)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~970368","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I973912f51f1846b1a8d0a2af5aa7a304a4c7a3a6","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"venv_package_list_conditions","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-03-09 10:46:34.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-03-09 13:20:25.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-03-09 13:20:25.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-12-11 07:18:21.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-03-07 11:17:58.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I973912f51f1846b1a8d0a2af5aa7a304a4c7a3a6","subject":"Remove group conditionals from venv_packages","status":"MERGED","created":"2025-12-10 11:59:44.000000000","updated":"2026-03-09 13:22:24.000000000","submitted":"2026-03-09 13:20:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970368-venv_package_list_conditions","meta_rev_id":"91c7898cc0f247764c85d6fd578f528d6caf8b27","_number":970368,"virtual_id_number":970368,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-09 13:20:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-03-09 10:46:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-28 13:36:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-03-07 11:17:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2026-03-09 10:46:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2026-02-27 16:16:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-10 13:52:52.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-12-10 14:12:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-23 18:05:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2025-12-23 18:05:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2025-12-23 18:05:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"3abbb2a16b2716e040e321c2219cf28000d6857c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-10 11:59:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3303704779f39b1c08c6550acdf172d09de84f4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-10 12:01:33.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"c6491d0cdd9e5682aac41d9dabaeff36da6dbfba","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-10 12:39:45.000000000","message":"Topic set to venv_package_list_conditions","accounts_in_message":[],"_revision_number":1},{"id":"61a49c2a3d5f8b8e373c6b7a0950171c5d16cb43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 14:12:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9856595a31e74ea79c28f74753863630\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86ec7873815a44dc8719e7e3b880a5c0 : FAILURE in 5m 41s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/9e196b4084a9440a8f27c72517127168 : SUCCESS in 41m 31s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/a33878a918954340ab83aa625f4c294c : SUCCESS in 1h 36m 50s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/bd0d22b7a9cc40fba2533d4630b36c6f : SUCCESS in 1h 26m 02s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/06bdc64d5a2740b68bc01e055a3a347a : SUCCESS in 43m 05s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5fd1e2c10cd84ad0890667a92a2da739 : SUCCESS in 44m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd9ca661fc4649908b7cf3c827cf89b9 : SUCCESS in 1h 22m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e06cc466c3a74345bad50667e45b5f2c : SUCCESS in 1h 40m 31s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88e3471e2fec4defbc898b4c42e8bfc9 : SUCCESS in 1h 58m 47s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36ca8ace3b4049a3948692b4a546ba3b : SUCCESS in 56m 03s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/478808cc1ce2446babfa5a69083750b2 : FAILURE in 10m 51s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49dbdfd21c044b7191208da82bc55bf0 : FAILURE in 53m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/968635825afd4ffe955710bdfd524998 : SUCCESS in 5m 25s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/232339d425fd463e8c8991df93b4c548 : SUCCESS in 1h 01m 21s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbcbe9e0bf7342a68fcfd300219d653e : SUCCESS in 1h 39m 46s","accounts_in_message":[],"_revision_number":1},{"id":"5cd3b88b6ee7087a7e6853e82dac32d4cbaf1422","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-10 16:19:24.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d866856e180d876f9d0912715341abf20fc42698","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 18:29:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2b6b08a36074bb98844dc8a2afcbf60\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01c536cf23ec40e690f2d9c493a90fb7 : FAILURE in 4m 50s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/b46ec84ab5ec4cdbb062ee47d2644048 : SUCCESS in 42m 15s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b0b9440d7bb44642ac3e3651a0879060 : SUCCESS in 1h 26m 31s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/d531edad90c74bcb8612aa13cb40629e : SUCCESS in 1h 25m 35s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c0d32b29d8bd4042b4bc69f71b7b9039 : SUCCESS in 1h 17m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/14192660bebb4b2b8fab751a6a71d250 : SUCCESS in 1h 11m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39af90c2fe944a47beaa1512006ea223 : SUCCESS in 42m 50s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98544f3aa9834148b2492d63f91edcfd : SUCCESS in 1h 36m 17s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7589548dee14ae9bb1e04b8fec1e3d8 : SUCCESS in 1h 00m 13s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f78008a870ab40f480dc637af4f4390d : SUCCESS in 2h 04m 13s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/64894a446d6147bfab6295a103f4df80 : FAILURE in 10m 45s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3231a7e36d404fd187870848fae6e58d : FAILURE in 53m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e9d3c9b54be40869e2a2ce4d36b763a : SUCCESS in 5m 42s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c176455d55b43a0b664b5a589f96147 : SUCCESS in 2h 03m 25s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4da3ec8d075648ef8dc563bfc076b2d9 : SUCCESS in 1h 27m 45s","accounts_in_message":[],"_revision_number":2},{"id":"ea1e4758223324b1738cd3ed40c83597e7cfc051","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 07:18:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8456590f643cff6ed7427d2ffad5a5310f80fa81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 09:39:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6abfec6c19584a0795807cf71b813ef9\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e293c7ac3d7f45c49f1a59473420e097 : FAILURE in 6m 36s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/306fafc9a11b482fa24d362ccbc48b88 : SUCCESS in 1h 01m 29s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f0b0231bafd14d50b589fb500913e555 : SUCCESS in 1h 26m 07s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/eaadbee4fbc44908beab8060ce0d89e0 : SUCCESS in 48m 34s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3620c69b52f0410491669d5748e8fd44 : SUCCESS in 46m 34s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ea74990356724ac9912f81dc6ed2e679 : SUCCESS in 1h 11m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00a12e4e85b54a6a8b70efa220bbce8b : SUCCESS in 1h 02m 50s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/daaf4a85964e4d558c4e77f657552286 : SUCCESS in 1h 29m 49s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a30082719f44f9b9ddb5044d4b697a0 : FAILURE in 2h 04m 35s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f169129dad14150a88ff609e990d4a4 : FAILURE in 2h 09m 09s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c7d953f5f8aa4e798bafbb449245788d : FAILURE in 7m 06s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54b4798fbbea42d89b66abf8e7db8d36 : FAILURE in 54m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/990acb41d9d340e0b90d450b6020b746 : SUCCESS in 5m 46s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfbeeeaedd7c4a23992650661dcae3bb : SUCCESS in 2h 14m 17s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb3ce26e00b74edaad40f46543b67423 : SUCCESS in 1h 21m 58s","accounts_in_message":[],"_revision_number":2},{"id":"5c45d67e74e1318d5cd9dab1e902455bf2391fec","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 23:54:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"57970f91d4a4c6665bdd2a061d405df09760da6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 02:13:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9146ccfda35742d0887872580662a25f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ea16a3df3ee4469804e75cbb5f9b76f : FAILURE in 6m 38s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/67d0d643bb78477e92872184608afe41 : SUCCESS in 1h 05m 32s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/48e3516a64564104be3016fceab45785 : SUCCESS in 40m 58s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/dde3a40129bb4253958823cb734ab931 : SUCCESS in 1h 16m 06s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/64a8c158d40646adb5d14f2f7142e70e : SUCCESS in 1h 15m 36s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d3737c58cde44dfc90fabe730828bb7e : SUCCESS in 1h 10m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6e7ea78dedf43b99c863e2a31aec56f : SUCCESS in 1h 29m 16s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec5396debca94967b2ac2fac1ea2b7c1 : SUCCESS in 1h 27m 01s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6416e8ded8a47668a52ed8a6bd90403 : SUCCESS in 2h 04m 47s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a83559db96142c38feaa51d03132af3 : FAILURE in 1h 59m 16s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2c6d676c65bc4f23985d88360ba08a3e : FAILURE in 8m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48780405ad29403daeedc78859e65947 : FAILURE in 28m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a372396ded547b9a3494aed72e455e7 : SUCCESS in 7m 06s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91e730a4d0794ad0b99ca126aa591f4c : SUCCESS in 2h 06m 55s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65c0effb3ac741309c4de586b0c3027c : SUCCESS in 43m 03s","accounts_in_message":[],"_revision_number":2},{"id":"83f063b8284c9ca76c09d17d0b6ed4f4baf79529","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-17 10:56:43.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"bde8fd8bafa6df18edeb0aec14f7646215689c82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-17 10:57:19.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"305a0400f4fcbc6929b2fce863ba66c1543aea5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 13:06:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1197c6d9032643f0a41abcbfc3e5ef86\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba5b15d722b341a4b6a3ff9715e099b9 : SUCCESS in 5m 01s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/7adc77d3d82c487b84d81169ab69a012 : SUCCESS in 1h 25m 52s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b265b02a6d42427a9de3ae3345f66b88 : SUCCESS in 1h 27m 09s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/5cc45bb8346b4f0f91fd34955f1e2c03 : SUCCESS in 1h 02m 57s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/78dc6cf071724070ab3d5bf75a366dd1 : SUCCESS in 42m 11s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b8c8068c513349dfad2641e2c3e87576 : SUCCESS in 1h 08m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73452bdc818a4bf79a03b7d239665ffc : SUCCESS in 44m 00s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dafa4fffdfda488f9e1a46a29720ef80 : SUCCESS in 44m 41s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/acd04f8271354fee9a36de75479fbe84 : SUCCESS in 1h 37m 50s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/85be24c2fbc44feabd62485f07ffc904 : FAILURE in 32m 42s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d36a8d7cda964535af436c28d067dcfe : SUCCESS in 1h 45m 12s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92a133f67f7f47678f13ee8077b14f57 : SUCCESS in 2h 02m 36s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74d480812be0492c9bb79c3f8b03ee8e : SUCCESS in 1h 40m 28s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c0d2ef0d3e354456be9ff267d91abccf : FAILURE in 7m 37s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c30180af7ce04665a396f4e847e464ae : FAILURE in 29m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec4f1f0b69dd4fd09ec283f20447e312 : SUCCESS in 5m 51s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/308634285cad4ca3b116804983eb162a : SUCCESS in 1h 01m 08s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbf08461f55f4a4f90a6f91c6ab83f0e : SUCCESS in 1h 20m 16s","accounts_in_message":[],"_revision_number":4},{"id":"ee87be0069998d7f58c78abe4162154c278a4735","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-22 21:12:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"376def78e3d4a76a9386b65b6503e4b95a4ba9ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-22 23:34:15.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/47eb6f2a0a224ba5bd49cc9cfff8ceb2\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efa844b2d8384bd5b93f3e71713e620b : SUCCESS in 9m 51s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/d0ac503ef1cf4ae9b5b7b1931d86d466 : SUCCESS in 1h 28m 30s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/1994019ccc794ef98332c24dd3128164 : SUCCESS in 45m 11s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/b45946743bb941608a7436d9404b7ea0 : FAILURE in 1h 07m 14s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a89718f86c58406989cb6eaeb0a14983 : FAILURE in 39m 15s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/13a9d0b83352491cb6a9e362c4eae278 : SUCCESS in 1h 38m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b40fcfed4add47bba1a9e9a288c8b1b1 : SUCCESS in 45m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3718a299a88c4c6f8bb577d8c1acd736 : SUCCESS in 1h 34m 44s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c69ddd3b4a58411298a3f1ac182d566d : FAILURE in 56m 52s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f03c24e264204d4fb4888767f34af01b : SUCCESS in 2h 09m 35s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b7a79b5d95841edbd5f03d9693f01dc : SUCCESS in 2h 14m 03s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fcdacd1d38242a0867c80731c1a7483 : SUCCESS in 1h 57m 40s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91aa1e2a3c4c4c4ca839aeb80b631481 : SUCCESS in 2h 08m 25s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2279491919c941a3a1076c4a4c67d885 : FAILURE in 6m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3179830c6f534990829c854ae9d93d17 : FAILURE in 59m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b453bebf54bd434ca0558c6c697fe02c : SUCCESS in 6m 21s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55f0d6f571ba489388071c35a8530d27 : SUCCESS in 2h 01m 43s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6e02c01ebd64e1bb2ec996c048a6563 : FAILURE in 1h 07m 34s","accounts_in_message":[],"_revision_number":4},{"id":"f8205955d474a94b11dae0501d565d828d945106","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-23 10:29:32.000000000","message":"Patch Set 4: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a029b3a36d650dad5c0b94331691cb54e7460f1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-23 13:00:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f93bc82e1a0d49178281eed48cc99ee5\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc6fa684ff554c3092732d7257239899 : SUCCESS in 6m 14s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/951a853ac15c4e49b9be3daf629e4adc : SUCCESS in 1h 17m 50s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e3390e66325b464897b41b75c896908f : SUCCESS in 1h 20m 09s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/8164ccad11544bf29105f68d2f6fca79 : SUCCESS in 1h 14m 41s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d9e15501da204f7189490694447847cc : SUCCESS in 48m 27s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/289e3b5f7b2e4c0fa069ea9ae3ac3ebc : SUCCESS in 50m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8241f56b1984d6fa31fdae89f4c8824 : SUCCESS in 1h 19m 30s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b264ad10c630461dace130268870e96a : SUCCESS in 1h 26m 24s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1f6a21821a614480bac6475eb3244356 : SUCCESS in 2h 15m 41s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fad8d44c59c84468b78ff673ac3d36d6 : SUCCESS in 1h 42m 44s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68e53c0e610a4982b368404f485511ff : SUCCESS in 2h 14m 39s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ef8c0a560134cc08432882151124211 : SUCCESS in 2h 02m 01s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f39a71f54654ebda0a2b5cfedd5b1d8 : SUCCESS in 1h 05m 27s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5afb88e3d92b41818d30c6b432feb754 : FAILURE in 12m 02s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c33eb4b69204e8d95d7a81132b7c431 : FAILURE in 49m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bfdf19d0250489590d51be9d076d877 : SUCCESS in 4m 53s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2e11b5efd9f4913b75795d92486ad20 : SUCCESS in 2h 01m 58s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64a10579fb2741e4b865f272af78a791 : SUCCESS in 1h 11m 39s","accounts_in_message":[],"_revision_number":4},{"id":"b596919f34b12db529708cb03627147ce011dcaa","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-23 18:05:04.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"266bf1176c882f3c7ffc8be1f0810dab9e1c3737","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-27 16:16:19.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR **is:MAX** OR is:MIN\")\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"07fe0bea8ba00b16511a42b9229df5e05bf3c22d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 18:54:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/40ac73790bfd48d3a17272f5d4e6685a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec2494a064524e819edc1dd78d351f6b : SUCCESS in 12m 50s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/5b13178f264e49119fe51c51006c0d57 : SUCCESS in 49m 33s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/58d0d955b20b44218f0f08fcebb659fb : SUCCESS in 1h 28m 22s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/359a850c88764522902436afb4c355c3 : FAILURE in 1h 21m 06s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a9215d61c1324c3db3897aa9392e8e21 : SUCCESS in 57m 26s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a30f10df471c4e13a0f889ee66035c4b : SUCCESS in 1h 19m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/609743b83ed74b079de090443790374b : SUCCESS in 1h 33m 22s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/310b442dc9284bb4889fa5b514d49e0b : SUCCESS in 1h 38m 47s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/86ea633bcd674e25a11f8a6d34f80ae3 : SUCCESS in 1h 48m 55s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/33caa2b4d1c6457a85149dcda3d46a3f : SUCCESS in 1h 44m 09s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06ec0e2a42224824bc6acfbb2f60084e : SUCCESS in 2h 12m 39s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4cd865773f74b32844c21762c5b9033 : SUCCESS in 2h 07m 50s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46f0eb6578734d5fb421ed47dd1ef485 : SUCCESS in 2h 20m 58s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/18bdfa81c1bd4bd4b92b4a4209a081c7 : FAILURE in 8m 51s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3693f15656de40f095a6ee7b56e006d5 : FAILURE in 50m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6c2024daa2e458996776cc0a9fa78d2 : SUCCESS in 6m 18s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45a6a223470a44549849253d88f12a81 : SUCCESS in 2h 23m 26s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/931d7862714c4a2dbac6923ef4f097de : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":5},{"id":"17324b54fe40264e88425dfa2566641c4c104f39","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-28 13:36:33.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a813c58cc6322bacac606b7d951e318c72f0b3be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 19:51:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c34186bb97814b6fadc97f6f3f4e13a1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bd04ec9d5e84b20af822363b363fe4a : SUCCESS in 16m 04s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/c82a219729b548f293862a262da416df : SUCCESS in 1h 14m 57s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/1b70d7bea9004635a935d1ce6547071c : SUCCESS in 44m 55s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/f30906f25fcb4d23adc0a24433e84d7c : SUCCESS in 54m 51s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/397d648084804d6985251a8d2df0f157 : SUCCESS in 52m 43s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d7b6b408fcf1420eaec9663bbda4312a : SUCCESS in 48m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ee17cbb44534410ac4b8f8b8afa80a0 : SUCCESS in 1h 38m 33s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2c17bfe00a04710aa6003571d3e7c1f : SUCCESS in 47m 11s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e0d8026fcc1d4d4fac71075b57cad051 : SUCCESS in 1h 44m 05s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/740695328d2847c18aa8199c4df564bc : SUCCESS in 1h 47m 45s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5019bee100664a66a27b3c4931e2e47e : SUCCESS in 1h 03m 37s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e388e9a5ef08498c8766446464db1df1 : SUCCESS in 1h 48m 25s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/160be0b9b7c4465cac5052d1192d0231 : SUCCESS in 2h 29m 27s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d8dcbdcfbccf4d598eda38bfcf179940 : FAILURE in 10m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f38c2d9b3aa449c9894994363f1a887a : FAILURE in 31m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49dabb0ef9444098b8a3493c5d70ec75 : SUCCESS in 3m 42s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f25196a498b4413fba48092781413582 : SUCCESS in 2h 07m 19s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65fc6c00d0324ec591b4fd17a2724338 : SUCCESS in 52m 09s","accounts_in_message":[],"_revision_number":5},{"id":"b0495f907b3ec75001e7bff6f6938f478d04f10f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-03-07 11:17:58.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"246e31cecfa030820ef2b4e6ad3440a85f9a2407","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-03-09 10:46:34.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"36a7df97dce8ee63bba8d51f73d236e5a40835f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 10:48:33.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1cb92628816ae823f5a6e22070d235503bc5a83e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 13:20:25.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e9ac241ab9945029bd9f464b6201e79\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/798258bb9f7942208a634d8b41c3f768 : SUCCESS in 16m 09s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/fa9670e0f5934b77822cd10fed2ca6b0 : SUCCESS in 1h 36m 08s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/14e5ac9ec3b94c99bcfa333bbb9b718f : SUCCESS in 1h 23m 35s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/bfd7ae08de2a4dd0b7b1f2dc8954f5c4 : SUCCESS in 1h 29m 49s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6af9224789ff41d7935d01e2647f160a : SUCCESS in 1h 41m 22s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a6563723e51d40cdac5d47cb22faaa31 : SUCCESS in 1h 43m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14ab74a4cbb74649a272ec67ff769340 : SUCCESS in 58m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/824131685f454282bc6d51ab6f29b164 : SUCCESS in 1h 48m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4af3379d789b4fc8bc098d16bd61eff5 : SUCCESS in 7m 10s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01bcb175d58a47ea8e1baefdb20abd03 : SUCCESS in 2h 13m 33s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01d7fce6d5a946e990b23dbab3da0b85 : SUCCESS in 1h 31m 36s","accounts_in_message":[],"_revision_number":5},{"id":"fb410e5bacee3bbcaa8c28010a927cea982025e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 13:20:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"91c7898cc0f247764c85d6fd578f528d6caf8b27","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 13:22:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/240c55261d4b42e4bb3260723b80a8d7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/269f5665f04f4625a42e9318379b12f9 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0dc0bf2721e4eb3edd51a359cc40dcdccedb6dd7","revisions":{"37cfbab314a13af610cbc5362dfab92794c96a6c":{"kind":"REWORK","_number":1,"created":"2025-12-10 11:59:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/68/970368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/68/970368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/970368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/970368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/970368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/970368/1"}}},"commit":{"parents":[{"commit":"2d2202114949de36f9215f9ead44733f325639af","subject":"Merge \"Do not remove policy.yaml file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2d2202114949de36f9215f9ead44733f325639af"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-12-10 11:59:37.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-12-10 11:59:37.000000000","tz":60},"subject":"Remove group conditionals from venv_packages","message":"Remove group conditionals from venv_packages\n\nneutron_venv_packages is used as a base for wheels build.\nBy default, wheels are being built against first host, which is usually\n a neutron_server.\n\nIn this case, wheels for these optional packages will not be build,\ncausing installation to fail.\n\nWe either need to have a different variables for wheels build and\npackage installation, or accept installing un-needed packages for\nall venvs.\n\nChange-Id: I973912f51f1846b1a8d0a2af5aa7a304a4c7a3a6\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/37cfbab314a13af610cbc5362dfab92794c96a6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/37cfbab314a13af610cbc5362dfab92794c96a6c"}]},"branch":"refs/heads/master"},"e29828db39fa8d86e5622e6296cb5c94fbd088b1":{"kind":"REWORK","_number":2,"created":"2025-12-10 16:19:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/68/970368/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/68/970368/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/970368/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/970368/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/970368/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/970368/2"}}},"commit":{"parents":[{"commit":"2d2202114949de36f9215f9ead44733f325639af","subject":"Merge \"Do not remove policy.yaml file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2d2202114949de36f9215f9ead44733f325639af"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-12-10 11:59:37.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-12-10 16:19:24.000000000","tz":0},"subject":"Remove group conditionals from venv_packages","message":"Remove group conditionals from venv_packages\n\nneutron_venv_packages is used as a base for wheels build.\nBy default, wheels are being built against first host, which is usually\n a neutron_server.\n\nIn this case, wheels for these optional packages will not be build,\ncausing installation to fail.\n\nWe either need to have a different variables for wheels build and\npackage installation, or accept installing un-needed packages for\nall venvs.\n\nChange-Id: I973912f51f1846b1a8d0a2af5aa7a304a4c7a3a6\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/e29828db39fa8d86e5622e6296cb5c94fbd088b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/e29828db39fa8d86e5622e6296cb5c94fbd088b1"}]},"branch":"refs/heads/master"},"0c6ec3237d0a1871b2f6f397d4e4dd66847a4959":{"kind":"REWORK","_number":3,"created":"2025-12-17 10:56:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/68/970368/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/68/970368/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/970368/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/970368/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/970368/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/970368/3"}}},"commit":{"parents":[{"commit":"2d2202114949de36f9215f9ead44733f325639af","subject":"Merge \"Do not remove policy.yaml file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2d2202114949de36f9215f9ead44733f325639af"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-12-10 11:59:37.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-12-17 10:56:43.000000000","tz":0},"subject":"Remove group conditionals from venv_packages","message":"Remove group conditionals from venv_packages\n\nneutron_venv_packages is used as a base for wheels build.\nBy default, wheels are being built against first host, which is usually\n a neutron_server.\n\nIn this case, wheels for these optional packages will not be build,\ncausing installation to fail.\n\nWe either need to have a different variables for wheels build and\npackage installation, or accept installing un-needed packages for\nall venvs.\n\nChange-Id: I973912f51f1846b1a8d0a2af5aa7a304a4c7a3a6\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0c6ec3237d0a1871b2f6f397d4e4dd66847a4959"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0c6ec3237d0a1871b2f6f397d4e4dd66847a4959"}]},"branch":"refs/heads/master"},"44dada5b0016281dcfba32c7b61c9d0226f9209b":{"kind":"REWORK","_number":4,"created":"2025-12-17 10:57:19.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/68/970368/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/68/970368/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/970368/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/970368/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/970368/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/970368/4"}}},"commit":{"parents":[{"commit":"2d2202114949de36f9215f9ead44733f325639af","subject":"Merge \"Do not remove policy.yaml file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2d2202114949de36f9215f9ead44733f325639af"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-12-10 11:59:37.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-12-17 10:57:19.000000000","tz":0},"subject":"Remove group conditionals from venv_packages","message":"Remove group conditionals from venv_packages\n\nneutron_venv_packages is used as a base for wheels build.\nBy default, wheels are being built against first host, which is usually\n a neutron_server.\n\nIn this case, wheels for these optional packages will not be build,\ncausing installation to fail.\n\nWe either need to have a different variables for wheels build and\npackage installation, or accept installing un-needed packages for\nall venvs.\n\nChange-Id: I973912f51f1846b1a8d0a2af5aa7a304a4c7a3a6\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/44dada5b0016281dcfba32c7b61c9d0226f9209b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/44dada5b0016281dcfba32c7b61c9d0226f9209b"}]},"branch":"refs/heads/master"},"0dc0bf2721e4eb3edd51a359cc40dcdccedb6dd7":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-02-27 16:16:19.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/68/970368/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/68/970368/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/970368/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/970368/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/970368/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/970368/5"}}},"commit":{"parents":[{"commit":"182afa875a7e54b6130a694cb21f0a342a048ed9","subject":"Merge \"Remove vlan_transparent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/182afa875a7e54b6130a694cb21f0a342a048ed9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-12-10 11:59:37.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-27 16:16:19.000000000","tz":0},"subject":"Remove group conditionals from venv_packages","message":"Remove group conditionals from venv_packages\n\nneutron_venv_packages is used as a base for wheels build.\nBy default, wheels are being built against first host, which is usually\n a neutron_server.\n\nIn this case, wheels for these optional packages will not be build,\ncausing installation to fail.\n\nWe either need to have a different variables for wheels build and\npackage installation, or accept installing un-needed packages for\nall venvs.\n\nChange-Id: I973912f51f1846b1a8d0a2af5aa7a304a4c7a3a6\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0dc0bf2721e4eb3edd51a359cc40dcdccedb6dd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0dc0bf2721e4eb3edd51a359cc40dcdccedb6dd7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
