)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~993147","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I5a446b7420842b4074ba5efb6e338ffdc734ad8d","project":"openstack/openstack-ansible-os_neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-06-18 02:50:37.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-17 22:01:40.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-06-18 02:50:37.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-06-18 02:50:37.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-06-15 09:58:54.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I5a446b7420842b4074ba5efb6e338ffdc734ad8d","subject":"Remove broken bgpvpn support","status":"MERGED","created":"2026-06-12 15:44:56.000000000","updated":"2026-06-18 02:52:03.000000000","submitted":"2026-06-18 02:50:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993147","meta_rev_id":"d2653c7ce8a0f5d95775e34dd96e04f1a2a095f8","_number":993147,"virtual_id_number":993147,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-18 02:50:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-12 23:12:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-06-15 09:58:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2026-06-15 09:58:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2026-06-17 21:58:07.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-12 18:33:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-12 23:12:24.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-06-12 23:12:24.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-06-12 23:12:24.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-06-12 23:12:24.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-06-12 23:12:24.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"f089ca36d93e1bd89e1fc6989cca4cb8ebb8c5fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-12 15:44:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61eafcd3e9841785b92891cd426ed1745cfd601f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-12 15:57:10.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"0855f7a9937c517d25b29461dbe89eaf16ffe9bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 18:33:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba3f261258bd4fc595e5bd7a7ba8f215\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f6aaf713104436997f315ea21d836b9 : SUCCESS in 11m 51s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/8c29ae8fd190430bb44bdc01e32f5600 : SUCCESS in 1h 38m 34s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f303b5f72155484ba713eef530b5da51 : SUCCESS in 52m 30s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/b7ce68249e014949906c67903f346f6d : SUCCESS in 50m 30s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1e86df1585234fc285de61101fa81acc : SUCCESS in 54m 12s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9faa8e7a07054106ad0268f94d283e9c : SUCCESS in 1h 48m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a42d50fe8b9844079e38241f084a9ee8 : SUCCESS in 58m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/01289f1d4a084a309fa039fcbf71f627 : SUCCESS in 1h 26m 56s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b66c5a494a4e4e3b916a35b773e6303b : SUCCESS in 52m 43s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9ba79ff6ae814020b6db2822eac1fb64 : SUCCESS in 2h 20m 02s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0225e783611a4ae6823cdd19eb2e1e94 : SUCCESS in 1h 50m 08s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b59bc5e41a1a425ab84a2ad9ba39c8b2 : SUCCESS in 1h 46m 00s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b740c9dd010141e59195cb884fee66c3 : SUCCESS in 2h 11m 53s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c55ff5e06a5a408ab9022491c4a3c1b6 : SUCCESS in 1h 11m 48s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b1886d6abb44018ac42bed635487829 : SUCCESS in 1h 25m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/7dadf3854f3b46f2842e07f378bf754b : SUCCESS in 1h 24m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd2ff7e03e6d47fdaaa303f9d8e8b440 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f8fb9cdbac74e87b9700cd7f2d85dfc : SUCCESS in 3m 46s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a50e6a7a9dc84a2ab68344fafe4721e1 : SUCCESS in 2h 12m 45s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ade3eff9b78a4d69bccd4be19c7a2b8f : SUCCESS in 1h 37m 41s","accounts_in_message":[],"_revision_number":2},{"id":"c941594fa71ada40a4ef5a9ad1d25d3aca606470","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-12 23:12:24.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bb9fcbe6df46784ba06d8da3d519c91458f1000e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-12 23:15:22.000000000","message":"Patch Set 2: -Backport-Candidate","accounts_in_message":[],"_revision_number":2},{"id":"fbbeb17a6faf8f2f2b2d75494dc09698e2e53d62","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-06-15 09:58:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ae630c73318bc84c51e0763b09b01381be86f88a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 10:02:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"734c729a624991014ba8415af23ba2e5932caa2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 12:29:24.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-ansible-os_neutron/+/993146 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"ba0dbe2da68cccb85008592c0357f95da5681b73","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-15 13:22:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2299af1aa40864dc27c9dd8ff5569e793c57f4e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 16:07:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5c4f200110649b4933da6d6f67797d8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d88b57b691fb43079143a5b5a719518c : SUCCESS in 14m 05s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/2c4b05082d064de094280225d951e546 : SUCCESS in 1h 34m 19s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/21629d7f07a1432196afb8b503c201f8 : SUCCESS in 58m 49s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/5414e579ec564a8683a6cebddc4cb103 : SUCCESS in 1h 37m 05s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ff01aa51c9d743a9b9c8b19ec26dbf3c : SUCCESS in 1h 14m 10s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/73387cb92db348c6958033a3afd93387 : SUCCESS in 1h 22m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49648314ae824a679112b8f50abf7a5f : SUCCESS in 1h 21m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/14bf2ccb5d454779ac6563adb93d0d14 : SUCCESS in 1h 00m 26s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34f293a3982f47f2a90ce0fa4c96ca63 : SUCCESS in 1h 42m 00s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/989cd2f6c51440f38d6a24104eee2ded : SUCCESS in 1h 15m 50s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/aa960af854f7424780eebb20d593d8b5 : SUCCESS in 1h 17m 59s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1a9f5bee02942dd8bb244cc0238b679 : SUCCESS in 2h 30m 25s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49b4fe693b7c4490a9241b13c6870b2a : SUCCESS in 2h 30m 40s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23ef464193044d56a6bf3dc02fc774f1 : FAILURE in 1h 19m 07s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61569abfc6a04b12bd7633c0f0ed6ed3 : FAILURE in 1h 16m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/0795b90846554055992f049ce430bc88 : FAILURE in 1h 07m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4686391f43b45dcacffb38743423fa8 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f452d8f09ef4011b6eef526dd9ce82b : SUCCESS in 3m 59s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f39b0fc089d4499ca12aa3cc97a2a20f : SUCCESS in 2h 25m 19s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4536f226c70b4ac099689b51d7abba8e : SUCCESS in 1h 39m 20s","accounts_in_message":[],"_revision_number":2},{"id":"baf554dccb48fd7cc33679ea44baabfc23e3e350","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-17 21:58:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b0888eda8f198c9c60d74a568e35343da1f10a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 00:20:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/779f0eef36c945f6adc8dd717cd56ef5\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4f952f456f4477198508e106c8ee927 : SUCCESS in 12m 13s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/546ce2bb516c42659f65083cba01f428 : SUCCESS in 1h 18m 07s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/48f0cc277e8d4fc0b8acacf1deb57b79 : SUCCESS in 1h 11m 28s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/4cd953271e6e438995b996827f73b9fe : SUCCESS in 1h 24m 35s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/71ada2a782fa4649a4f7db81d4af5c71 : SUCCESS in 1h 12m 31s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a0a1a8dc115947e7a8900080b9b9391a : SUCCESS in 1h 19m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9757cdc38094bfbaf0ca4282c4b4435 : SUCCESS in 1h 33m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d07d709cf06340d08b04feaa5a951893 : SUCCESS in 49m 23s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01ad1384c373444e9c1fc093eef32de4 : SUCCESS in 1h 35m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1ee4bfccb4b840eeb725fb3ec1aa257f : SUCCESS in 2h 17m 11s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3b81e033965e44abafda397d03b27093 : SUCCESS in 1h 48m 38s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9df3d4eacf5468cac9a2bc55f93d5a3 : SUCCESS in 2h 11m 25s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0a5f74336214df9bec1a8e07e323597 : SUCCESS in 2h 12m 54s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8070248c2fd4c0387c26e700c5b6ec6 : SUCCESS in 2h 14m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a167812830864cbc89e365e50dfe101e : SUCCESS in 1h 28m 41s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b7e6ab16fc22410cb941233f617c4106 : SUCCESS in 1h 25m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8df7508084824e7c888f45c7701503a8 : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2eaabb58dd964339b3485ec9fbb64dfc : SUCCESS in 2m 31s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0bc5d16ba0149e59e049951d3da0253 : SUCCESS in 2h 11m 29s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e481b1ede6e64360aa317fa95b7904a9 : SUCCESS in 1h 39m 22s","accounts_in_message":[],"_revision_number":2},{"id":"a3c1a0eaabc3bd87d5340d9e54eead470ea97467","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 00:22:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6d957b42d51bd14cac16eaa886e9296708d151a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 02:50:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54c123ef2b284914b2ab98c2ecd91e74\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/501f3eb0a5934ca29915e1c00e26e49a : SUCCESS in 10m 50s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/78a553a34ae14fa78d7bbb8883186aa9 : SUCCESS in 50m 19s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/6a403196fe7a4b808e383986474c23ee : SUCCESS in 1h 12m 00s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/870f284267a74a7f85804cd13f1b6125 : SUCCESS in 1h 14m 38s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c0d17d083b3e4379aa98030d382fcd03 : SUCCESS in 1h 15m 38s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6b42ae9a94754a399c2203a03bf4c1a9 : SUCCESS in 1h 11m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07703210ec3e4c0d9626641fb866b7b7 : SUCCESS in 1h 15m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/df51979e4d17407899eeeafa38198fae : SUCCESS in 1h 08m 03s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aab128bbbd9e4ce59aa4e17131f8400a : SUCCESS in 1h 32m 41s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec76ee8d8f96495ab3f4026150174718 : SUCCESS in 1h 07m 30s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/803b040dd03d43058c0902a12d5e2f1e : SUCCESS in 1h 24m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/146348045a3e4b5a9964e5f257f61d81 : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d1b0c850ece4e2da28632d3465dbb9a : SUCCESS in 3m 00s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a76dfdf8a63a458b9f1b0c1ed2f53c52 : SUCCESS in 2h 22m 52s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2f78de5a71844008e72a494f7ef48a4 : SUCCESS in 1h 32m 56s","accounts_in_message":[],"_revision_number":2},{"id":"4e9c46e1bff5a67d12c75e58853f6f86af58cbf9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 02:50:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d2653c7ce8a0f5d95775e34dd96e04f1a2a095f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 02:52:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb65d328d5ea4a4eb277d1f9df530546\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08df5d2bb5364e00b8c86136a3ed4339 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"40aaa65cc83e10e557a81d63b48292b83ab5191f","revisions":{"2470e96807ebb2cda183b53e16aff2164abb28f5":{"kind":"REWORK","_number":1,"created":"2026-06-12 15:44:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/993147/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/47/993147/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/47/993147/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/47/993147/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/47/993147/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/47/993147/1"}}},"commit":{"parents":[{"commit":"2f5092a12c0b15f303ab528d13f4fa95f9889ed4","subject":"Rremove unused ceilometer repository information","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2f5092a12c0b15f303ab528d13f4fa95f9889ed4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-12 15:44:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-12 15:44:10.000000000","tz":540},"subject":"Remove broken bgpvpn support","message":"Remove broken bgpvpn support\n\nThe current installation of bgpvpn requires networking-odl, which is\nno longer supported.\n\nChange-Id: I5a446b7420842b4074ba5efb6e338ffdc734ad8d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2470e96807ebb2cda183b53e16aff2164abb28f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2470e96807ebb2cda183b53e16aff2164abb28f5"}]},"branch":"refs/heads/master"},"40aaa65cc83e10e557a81d63b48292b83ab5191f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-12 15:57:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/993147/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/47/993147/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/47/993147/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/47/993147/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/47/993147/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/47/993147/2"}}},"commit":{"parents":[{"commit":"0d7fe265b2c9bf6f2bb8bd305ac4f337fe1a3cd4","subject":"Remove networking-odl repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0d7fe265b2c9bf6f2bb8bd305ac4f337fe1a3cd4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-12 15:44:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-12 15:57:02.000000000","tz":540},"subject":"Remove broken bgpvpn support","message":"Remove broken bgpvpn support\n\nThe current installation of bgpvpn requires networking-odl, which is\nno longer supported.\n\nChange-Id: I5a446b7420842b4074ba5efb6e338ffdc734ad8d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/40aaa65cc83e10e557a81d63b48292b83ab5191f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/40aaa65cc83e10e557a81d63b48292b83ab5191f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
