)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~994150","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I663915c70c8eb55931df9317b05ce89157629e93","project":"openstack/openstack-ansible-os_neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-09-28 09:21:51.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},"13095":{"account":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"last_update":"2026-09-30 11:58:41.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-09-30 11:58:41.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-08-31 07:00:14.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-30 11:58:41.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-09-30 11:58:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I663915c70c8eb55931df9317b05ce89157629e93","subject":"Add pipefail to neutron-l3 cleanup handler","status":"MERGED","created":"2026-06-19 18:22:08.000000000","updated":"2026-09-30 11:59:40.000000000","submitted":"2026-09-30 11:58:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994150","meta_rev_id":"4d08463850890ace923b229a6f30c8e99ec70519","_number":994150,"virtual_id_number":994150,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-30 11:58:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-09-28 09:21:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2026-08-31 07:00:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2026-09-28 09:21:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-19 18:25:56.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-06-19 20:56:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-01 16:14:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2026-07-01 16:14:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-07-01 16:14:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-07-01 16:14:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"b3025454239fe9f7b83bdf21c620fba3d54bf4ee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-19 18:22:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e761de5508c726bda5e168925368ed907e79fd17","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-19 18:25:49.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d998b4c7880729c52d1f5c163e5d512b0a280f10","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-19 18:25:56.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"e4dfa938752f92a723095760242ba44941621b17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 20:56:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6124fd420e484dbda9dc19005173b0c5\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b932d306c64441099b61650f211fc8d : SUCCESS in 10m 00s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/ef362d4951d1497b91c4c0affae947b5 : SUCCESS in 52m 45s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/c09781fb745c41b385343ce74ee19cc4 : SUCCESS in 1h 33m 13s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dbb8b0c431444a27bfd5b7f4a9d4e193 : SUCCESS in 51m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/db0a16c399e345b1a563f9786b4d0b3d : SUCCESS in 1h 15m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd1fb0438b2b47ed92a7c3de302ebe9e : SUCCESS in 1h 14m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f937d09160f849809f5f18c792c65651 : SUCCESS in 1h 29m 27s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89e0697ea8da45f88a51b6289c119a04 : SUCCESS in 1h 24m 17s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b02b376fef2c48ea83089712568c7b59 : FAILURE in 1h 49m 33s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e3b9f6691f59460eb4c36d2d3adb52cb : SUCCESS in 1h 59m 34s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b57e81f8c13f45448e79bd88faa22647 : SUCCESS in 2h 16m 57s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf856b625d6b4316bc2ad396872b6437 : SUCCESS in 2h 07m 26s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50983f7ed8084843b97b5fe824a8e699 : SUCCESS in 2h 22m 34s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2bc5db7cefc4cac9659495546d5e389 : SUCCESS in 1h 25m 22s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/4227d6c4162a4b0cbbc6de8e31e7552f : SUCCESS in 1h 07m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc0d82b4bd61461395d6a282e201593a : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7597670fa084f0680283ccd9a7e98df : SUCCESS in 3m 24s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c3c12df364d48978690f97a5589f600 : SUCCESS in 2h 14m 20s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/905bcfe99d3b43c0991a41471a04fbc1 : SUCCESS in 1h 26m 11s","accounts_in_message":[],"_revision_number":2},{"id":"a088740c6553435702e61fbfb0dca9ee4fa4e9c5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-19 23:15:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d91dc9a6981d10150f529e49c2bf7434e93cbbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-20 01:44:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb6b0e835b274745abd6d2c8b0689a35\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e00027289394e1e906e7a483619d01d : SUCCESS in 12m 09s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/900973bb99f44f55b4e0e5f331acbc97 : SUCCESS in 1h 31m 39s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/be9c1a1fce8143ccb9efdfb6602dae38 : SUCCESS in 1h 27m 26s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/62ba240cc36a496daae664dfc0d8b176 : FAILURE in 1h 07m 59s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dfa481858f954bfcbb60bedfeb7bd57f : SUCCESS in 53m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/221e46b47b594fe8a6b41e19ae87ed70 : SUCCESS in 53m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/5891460bc24a4fe29fe37ec52bd47aa4 : SUCCESS in 1h 05m 12s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbb4133f42264a55bd883d6fffc0e63c : SUCCESS in 1h 09m 57s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d041ae6512844d8e8de1e6ba22024af3 : SUCCESS in 1h 48m 55s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3552034010eb49428e12b6c69be7445a : SUCCESS in 1h 44m 26s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa72acc3f6d749fcb3b718b5d444491c : SUCCESS in 2h 20m 30s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6cbcdd19c294599bd92880be6e1dd35 : SUCCESS in 1h 07m 37s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/718fcebaf02a477bacf5928feb5164a7 : SUCCESS in 2h 20m 07s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0574651f18f4f7ebafdcc1f880b6b4d : SUCCESS in 47m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e91ede33846b427eb0b82877ea87495e : SUCCESS in 46m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b91ee3add5c94fad853e33d14faef9fa : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32d24cb956564132aeed2cb306e6bbc9 : SUCCESS in 3m 14s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1438546a9ee7423fb803708b5c23ff41 : SUCCESS in 2h 13m 27s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2d9b9ef3cae48a8b9c8deea9ec76b09 : SUCCESS in 50m 35s","accounts_in_message":[],"_revision_number":2},{"id":"331ed0a3742d47111c8ac8da67bfba0280ca2402","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-20 08:50:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cbc0f7c297976a76633c43c15d6d267b651f7f17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-20 11:18:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea40ecae79bb4f0794c9eeedbf20a089\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25c647ddf35f4e639aa13abd83ae1824 : SUCCESS in 9m 46s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/310b8418825745a7b0685b981034fea6 : SUCCESS in 1h 07m 21s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/e5cfbfb906624c028b95d00bc8938530 : SUCCESS in 52m 32s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0337af8997674f339b8c269c56e9ecaa : SUCCESS in 1h 33m 07s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0782e6e2ae30497b9af61854d375ef31 : SUCCESS in 52m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3aeb6dbbbfcd4a93896693f3b23f4084 : SUCCESS in 1h 06m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/4a6e8c7e5d634a03af6fc2b04d5ec8b5 : SUCCESS in 1h 13m 10s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a82fe51d42c5466c8220f14f7fb7462e : SUCCESS in 56m 18s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/046969a8d61a4025bca99805687d914e : SUCCESS in 2h 13m 22s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/37caaa84f95d4cc9bc5db8fd31718317 : SUCCESS in 2h 19m 11s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d87eb87ae84d4caaa4176ae9ab1b57d1 : SUCCESS in 1h 05m 23s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62d799fd681349c799bdf5b668165234 : SUCCESS in 1h 08m 03s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd449c192132472fb83ec884df0a38af : SUCCESS in 2h 15m 58s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/799ea29c0cd64228aeebbc6c579d383a : SUCCESS in 1h 26m 18s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/0ec9e0eb362348b89b8451e1268075b9 : SUCCESS in 49m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8640a9c0e38a4da6824abfffa14a2139 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54584a6b9d4648b9835aa448772dba8a : SUCCESS in 3m 20s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c217ba7c1f145159e13a07d8b4d87ee : SUCCESS in 2h 18m 06s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ba7a9cdcec74939b544788a8c29c4c4 : SUCCESS in 1h 34m 43s","accounts_in_message":[],"_revision_number":2},{"id":"6848d687340886ee014e91d82d75a55eebf88b8b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-24 06:50:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a46aa53eb24a5f0d9f87ca82045aff8da6c7111f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-24 08:33:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bec494df43d4eb42cdeff1cf206d2f747e0742e1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-01 12:36:53.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6caa140fbc9d03b65a49c792962e156993386455","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-01 14:20:23.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"e46625c574b95f3a90dcb3c43e4faeb721c7f8f4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-01 14:22:23.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"efeba5c8f61b67bb7d5b6750cdb59fd627336882","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-01 15:41:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ec332c00ce82d998d5a950c9e3d6a647759efe45","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-01 16:13:51.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"6c4c4348baef76bd5187404859bb2a2dcda13d81","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-01 16:13:58.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"e4c6e7e1c70fe372019fa4b9bd1aa1fb365a02e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 18:47:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/756b002ea774432ab9bdf6fad662abeb\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b73334187504865a0b3b4231c1dd0e6 : SUCCESS in 9m 26s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/a8c4313ed8e24c029d838bd49f6154d1 : SUCCESS in 1h 24m 39s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/79c14a7b7c8c4c238d652b0da1311c05 : SUCCESS in 48m 23s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1938d7d38fd74d07b346e9979f7599f4 : SUCCESS in 51m 25s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/029356dc693b4f14b8870b43b8ceae7f : SUCCESS in 54m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0b81e881a07479cba3fe72db41a7a2a : SUCCESS in 1h 37m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/9c44e35c2a46438c95148079043748d8 : SUCCESS in 1h 35m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ea03d7e8779445f9c1c8e91239e244b : SUCCESS in 49m 52s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f21c3ed172504437bc2063aee52c4926 : SUCCESS in 2h 02m 16s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfb1c97ab4384e1b9a4b8d46143fc710 : SUCCESS in 2h 26m 00s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60361049ca63463e9c03c9c4eb9b2f57 : SUCCESS in 1h 18m 41s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc79093345a84663b08ea9f7d5201401 : SUCCESS in 1h 13m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/627dde925085454386a4e8f54acbdf97 : SUCCESS in 46m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4772c238a4f49e7b19ddcdcc9dca323 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4cd5a6e168a4658bf8d43f4afb51901 : SUCCESS in 3m 19s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4495b705906c4a00b502c83f6535de7e : SUCCESS in 1h 47m 05s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c2b7d33b78c4d49acde05d036d6f1e1 : SUCCESS in 49m 44s","accounts_in_message":[],"_revision_number":3},{"id":"952f916b4c974e07bfc51ef2e0283c9ab1ba20d0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-31 07:00:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5caf7663a8532cd0b1c790bd1805dbb45a010c4d","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-09-28 09:21:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"53c6a188c46a96aa12715eb3b40371d293c872ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 09:29:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e69c99767146b27b8ce7eb0c28a170bae9e61bbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 13:19:19.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff4170d67d6143168d5d78e1041b474e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1aa037029a4a4895a653eed7d4810cda : SUCCESS in 20m 08s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/26c58cfe05e14aafb17148811f2b0154 : SUCCESS in 1h 31m 22s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/e498da01e96f4b45919d576d1dc46889 : SUCCESS in 1h 33m 14s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/74edda167ccc4090b37a58972f174708 : FAILURE in 57m 53s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9a77caf0296e43cebdba9e2e9e47cd43 : SUCCESS in 1h 42m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50b70420cf0f45c29e18eba3cc3b0246 : SUCCESS in 1h 19m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/de71cdf056644fd38d8679873cb39756 : SUCCESS in 1h 34m 42s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0dc35fb4e4434b1288dadb1794271135 : SUCCESS in 1h 38m 59s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4b3efcd73e9406aaf519588a267dd79 : SUCCESS in 1h 24m 28s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/34ddc835fa314739b20dff549915a021 : SUCCESS in 1h 33m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/951bea9a3c8a4a5185f747b2584c1105 : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/664f479d2be34ad08c147828e934849d : SUCCESS in 5m 13s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5861d7de11bd4c73845dfabb7c204ed2 : SUCCESS in 1h 21m 35s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2840fc6efced4cd79b85b8bcf3886c3e : SUCCESS in 1h 37m 49s","accounts_in_message":[],"_revision_number":3},{"id":"99ef5d2c1e93fe801bba249d9f945b0a9eb88133","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-28 14:47:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"299b65bd46a306136d418b3ea789c1f1a2088289","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 18:24:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c44e000e72d14cb9aef4cbb1d2200501\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/051462469d2d45bd9de48330dcdbbb09 : SUCCESS in 17m 06s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/2a5c80f64b6f402797362c695cb0603b : FAILURE in 20m 38s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/993de4d1e5f24c2fbac6adcbbd41eddf : SUCCESS in 1h 35m 00s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1e4806fc2c3a4ad6aff152fc67b5035e : SUCCESS in 56m 12s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3d8ecce2a7a54583bf71a7e32e1d3998 : SUCCESS in 1h 38m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01ddbe36aed24164a4079c642dc2d1b0 : SUCCESS in 1h 25m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d44942c982e1433fb11387dd85b534c6 : SUCCESS in 52m 22s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2687d9c5f982497fb8af39d19d490f80 : SUCCESS in 1h 43m 05s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/43dbfeb0fcc744dfb402ca2700d00092 : SUCCESS in 2h 32m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b95334a50c154f2a91f6b3661fc994bb : SUCCESS in 2h 21m 58s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0741bf0b6b746da97091db6f5a081f1 : SUCCESS in 1h 25m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/982de37a568c4d90bc0b167458bf8798 : SUCCESS in 1h 31m 38s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/8ed4b068c6714e5cbf055421d207144b : SUCCESS in 57m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2302a76656e49858d51d2ff2e9d52f3 : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fab1781491b4785bff8b7cd9aab01b2 : SUCCESS in 4m 09s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e3bd7957eba406784eaf2d4edffed4d : SUCCESS in 2h 22m 47s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c66f06060384782b6d9dfd512599713 : SUCCESS in 1h 40m 29s","accounts_in_message":[],"_revision_number":3},{"id":"cf455f48c8516757b83870687d465fa85fb3262b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-29 11:35:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c83b607af43d02e9635239b891d30b8dfd80637e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 14:53:46.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/678779c2f8e54f1ab554dfb8df9540bb\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac1e9a0252b74d9b81ed4ba7ae4e79c7 : SUCCESS in 17m 03s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/ea4e9e83464342b59aa76446ca9977f2 : SUCCESS in 1h 17m 58s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/ae4d20a2c4a84b0097c0f7a66a5b60d4 : SUCCESS in 1h 36m 32s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/19b489d2c08c438bbd7ad20e137b4735 : SUCCESS in 1h 26m 38s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4d3262a7d05a43a78a72a676c643ac98 : SUCCESS in 1h 22m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5cbd6b32c374ed28bcfc728a35c31b0 : FAILURE in 1h 37m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/4b7e80936aba4549b99f802093240a49 : SUCCESS in 1h 33m 12s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2774a8dab2a44509aeb8e23f65a96d6e : SUCCESS in 1h 42m 45s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/77c7c37bfe7747b4909e412f83697b61 : SUCCESS in 2h 31m 51s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e64d19d892054a5fb01cc1c79b4565b4 : SUCCESS in 2h 32m 49s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cfb26f7c87047fabf54cc03000b5ff0 : SUCCESS in 2h 37m 26s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ebc983c40d14bbfa085e24c08fc4f18 : SUCCESS in 1h 38m 56s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/918789e349224020803211042d67ede4 : SUCCESS in 1h 30m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b67c2ead1a1b48fca8741595022d2f5d : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b86cccea3d874918abe76329afff33e3 : SUCCESS in 3m 56s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7b80a745de4476e884071979f3c092e : SUCCESS in 2h 28m 35s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17e1d53f4c52413ea005bc2846b71993 : SUCCESS in 1h 39m 12s","accounts_in_message":[],"_revision_number":3},{"id":"7ba800153d83e12cbcac8a97c26cc06b51e0d57c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-30 06:56:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"037852aeaf6f55c27a975f96d8be72168b3fbe0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 09:49:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4d1bac84d8e4b93a7c28b18c41fdb63\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9630886327a44860b19a9bfa2ffc3fa3 : SUCCESS in 13m 54s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/6a6e7058b79d4e9fbb38937711f028b7 : SUCCESS in 1h 07m 34s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/d069955cb3c54b83972d4a2d075c3adf : SUCCESS in 58m 45s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/54ad414c42244c86917a39ea8a9f4cc5 : SUCCESS in 1h 34m 21s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/801c582345314dabb66d88e79263a0c2 : SUCCESS in 54m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b0717911c644b56a321169506104d83 : SUCCESS in 1h 32m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/00827b6dcf9043629a0f65533861ed09 : SUCCESS in 1h 30m 54s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a19c6c213fd748ab82e53ba2a7538b6e : SUCCESS in 1h 44m 05s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/786727e92dc04af0b610ffddd67ff5a1 : SUCCESS in 2h 04m 05s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e26f2953285d4bf1a0111881ede0cbd9 : SUCCESS in 2h 33m 54s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/368b9ae7dd3d44d38875fda9a4a0c900 : SUCCESS in 2h 38m 58s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/844f81be4b9e42d3ad022517e72997cc : SUCCESS in 1h 14m 31s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/a16c8e1d62324eb78f975feec9e76490 : SUCCESS in 1h 24m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f03b94c56aa34f19a5d6766d6c87868a : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/281116319fa7410f8993dc3dc066c004 : SUCCESS in 3m 22s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac4d4b669ede41cf92bdab503351adad : SUCCESS in 2h 15m 49s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fd71eaac45845a9979478c28ff0a8f4 : SUCCESS in 1h 37m 57s","accounts_in_message":[],"_revision_number":3},{"id":"47a080efddcad20616b8406da659ae3b6d00b577","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 09:51:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8e63079dd2527b06c535038d60eed46b4023abe1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 11:58:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c5366cb9e9941d9b7426e6bed3f50fb\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85312deb63d3405a9b358f918e0f30b5 : SUCCESS in 14m 46s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/354260348ff24fd5848afe5d9af4269b : SUCCESS in 53m 49s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/2c64490aa72d4833a8219ecdd5b687d4 : SUCCESS in 1h 33m 18s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/783d86da37d840b49063043debecc969 : SUCCESS in 52m 54s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9e07b03741094d1a85e06c6d02a283ea : SUCCESS in 1h 40m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f391b386cf814d4382558d9e2a7fc716 : SUCCESS in 1h 31m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/96dda6e1d5cb4649bf6a0084cdff538f : SUCCESS in 1h 23m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c11b56c32d04ad59710df561e1bbd6d : SUCCESS in 1h 39m 56s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd61a4bc903d4956b8250b3762f7bb7d : SUCCESS in 1h 22m 46s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/46c1afd5a9324799a63a9dd291f25025 : SUCCESS in 1h 31m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f025df3938924f8daf70b00659ba47dd : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e56d4e5f3fcc4c4a9e032926278fec0a : SUCCESS in 4m 19s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d61f9e74b1e04ad8b2352fd07c428b3f : SUCCESS in 2h 01m 47s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c25bc49ff8954aa88263506539945ac1 : SUCCESS in 1h 34m 28s","accounts_in_message":[],"_revision_number":3},{"id":"65728e082960818c5caab9f5534a84742243da1c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 11:58:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4d08463850890ace923b229a6f30c8e99ec70519","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 11:59:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d2a73cbe53c428fa50a5e4b44677b18\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e1afb4d9e064ff9b7f27fe02012ee60 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"664fc165cbaa84e3d4ebe35d1b0b7207a14acc8b","revisions":{"4267e626231dad86812d32a87c15c8d5bde33645":{"kind":"REWORK","_number":1,"created":"2026-06-19 18:22:08.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/50/994150/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/50/994150/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/994150/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/994150/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/994150/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/994150/1"}}},"commit":{"parents":[{"commit":"db0f3240fbffd5cee2ed6c3cf17bdecf73987543","subject":"Merge \"Ensure leader_node conditions are proper booleans\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/db0f3240fbffd5cee2ed6c3cf17bdecf73987543"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-19 18:22:08.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-19 18:22:08.000000000","tz":0},"subject":"Add pipefail to neutron-l3 cleanup handler","message":"Add pipefail to neutron-l3 cleanup handler\n\nChange-Id: I663915c70c8eb55931df9317b05ce89157629e93\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4267e626231dad86812d32a87c15c8d5bde33645"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4267e626231dad86812d32a87c15c8d5bde33645"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"07400aeb8dda2469f58bc1b8cdfaa78c30f26018":{"kind":"REWORK","_number":2,"created":"2026-06-19 18:25:49.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/50/994150/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/50/994150/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/994150/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/994150/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/994150/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/994150/2"}}},"commit":{"parents":[{"commit":"db0f3240fbffd5cee2ed6c3cf17bdecf73987543","subject":"Merge \"Ensure leader_node conditions are proper booleans\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/db0f3240fbffd5cee2ed6c3cf17bdecf73987543"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-19 18:22:08.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-19 18:25:48.000000000","tz":0},"subject":"Add pipefail to neutron-l3 cleanup handler","message":"Add pipefail to neutron-l3 cleanup handler\n\nChange-Id: I663915c70c8eb55931df9317b05ce89157629e93\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/07400aeb8dda2469f58bc1b8cdfaa78c30f26018"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/07400aeb8dda2469f58bc1b8cdfaa78c30f26018"}]},"branch":"refs/heads/master"},"664fc165cbaa84e3d4ebe35d1b0b7207a14acc8b":{"kind":"REWORK","_number":3,"created":"2026-07-01 16:13:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/50/994150/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/50/994150/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/994150/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/994150/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/994150/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/994150/3"}}},"commit":{"parents":[{"commit":"db0f3240fbffd5cee2ed6c3cf17bdecf73987543","subject":"Merge \"Ensure leader_node conditions are proper booleans\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/db0f3240fbffd5cee2ed6c3cf17bdecf73987543"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-19 18:22:08.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-01 16:13:50.000000000","tz":0},"subject":"Add pipefail to neutron-l3 cleanup handler","message":"Add pipefail to neutron-l3 cleanup handler\n\nChange-Id: I663915c70c8eb55931df9317b05ce89157629e93\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/664fc165cbaa84e3d4ebe35d1b0b7207a14acc8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/664fc165cbaa84e3d4ebe35d1b0b7207a14acc8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
