)]}'
{"id":"openstack%2Fopenstack-ansible-os_nova~1003252","triplet_id":"openstack%2Fopenstack-ansible-os_nova~master~I45ec88253aca9bf98227d8c783349e3cf7e1dd3c","project":"openstack/openstack-ansible-os_nova","branch":"master","attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-01 10:54:36.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I45ec88253aca9bf98227d8c783349e3cf7e1dd3c","subject":"Fix typos","status":"NEW","created":"2026-09-01 10:53:30.000000000","updated":"2026-09-07 02:24:37.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cca6bdf8bb8b7770e54ebe2f0a514132dadd57e7","_number":1003252,"virtual_id_number":1003252,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-07 02:24:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-01 10:54:36.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-09-01 13:39:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2b7a2c11e5db75bb3035876eba517005d21af095","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 10:53:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ddb49cdbd6b2bb87a7bd56f101f719ca10bc5e32","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 10:54:31.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0b74cae7aaa4415f087e7129dd20ad60890154d0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 10:54:36.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"50564b67863bb6893e3d45e8e8cc1ed7fa2df2d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 13:39:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b03d3c16cf8f4aeeaf128018e9403953\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40c7e196669a485786055b2cc370458c : SUCCESS in 13m 22s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/3403cd5e13fd4345bda321c77e7aab8d : SUCCESS in 1h 13m 51s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/ad27935a73404ab7bc8ef1efe87853cc : SUCCESS in 1h 33m 41s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fbed697a3d6c43c4b576b0d7ae6092bd : SUCCESS in 49m 52s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/66931ee0e0964f48bcaa506a9fd54aba : SUCCESS in 1h 35m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/760c13e11a2c426a8b59e6430b684e68 : SUCCESS in 1h 28m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/4507f2e2344e46c4877620f04c43b698 : SUCCESS in 1h 38m 27s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdc6e660592347fe87ffd015490cb01d : SUCCESS in 1h 34m 32s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/351b6cb42d1d43abb4aedcd96562d49a : SUCCESS in 1h 23m 51s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0861fa38e6a5480b8ca5e72b9382e4a3 : SUCCESS in 2h 31m 13s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8527173126224c83b832f3b7b8b14d08 : SUCCESS in 2h 20m 46s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4a7c05522cb47199d94ebc15616446e : SUCCESS in 1h 28m 53s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/53f47a9321c541c3b37aaf71ac65c2d0 : SUCCESS in 1h 30m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4f6512cba304b2d82b7be6a66993dea : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f16106c5d10461cae4449f055cab3cc : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":2},{"id":"3a0b0e9544441ce60fb2bba9c193afafbf2ba886","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 23:50:25.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"cca6bdf8bb8b7770e54ebe2f0a514132dadd57e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 02:24:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7aa57f62871146f0b907afb984a648f4\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/029d13959ba640d2ab728220f7313530 : SUCCESS in 12m 11s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/257a47af0b8e40528044662b9548e828 : SUCCESS in 1h 05m 27s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/8c65a6f9eaf147198712e47680c7554f : SUCCESS in 1h 05m 57s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f62fe0a09b88446f8dad22dfc9e08a6a : SUCCESS in 1h 14m 18s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/22a49d830e004e908bc13f44f67d663a : SUCCESS in 51m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d24e5dff7e04d69a0cf90be92e83a59 : SUCCESS in 46m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/9f0c17db33684db8a938853adebe0842 : SUCCESS in 1h 21m 43s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6552aedaebba41d09591a3c937ac91bf : SUCCESS in 57m 40s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c6bc9d4297dc4d3b8433c5f1cb083074 : SUCCESS in 1h 56m 32s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cb158487cda4c58a82adce692bcd51f : SUCCESS in 1h 22m 59s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/515366140e634218ad9aad5eafdcb52c : SUCCESS in 2h 25m 26s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9def651d18cf4ad6b4c122c2bb0b8218 : SUCCESS in 1h 30m 45s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/7886c1bae275468ba9dd37c015d12451 : SUCCESS in 52m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bcf3fcc9e0c410caede9a6ee5b43dad : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6f91a573862415da032a4acb44d9339 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e146401bc7784023c1c15ff8c680d29de2acdbcc","revisions":{"19579b091c9f91904deb093f01294381a9de90b3":{"kind":"REWORK","_number":1,"created":"2026-09-01 10:53:30.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/52/1003252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/52/1003252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/52/1003252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/52/1003252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/52/1003252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/52/1003252/1"}}},"commit":{"parents":[{"commit":"f427fbaa343ef1e6a96c9a64706e2124975ad0a3","subject":"Update ansible_managed variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/f427fbaa343ef1e6a96c9a64706e2124975ad0a3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 10:53:30.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 10:53:30.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I45ec88253aca9bf98227d8c783349e3cf7e1dd3c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/19579b091c9f91904deb093f01294381a9de90b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/19579b091c9f91904deb093f01294381a9de90b3"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"af684e8f0308c872173d6e5403252e4d1dc66349":{"kind":"REWORK","_number":2,"created":"2026-09-01 10:54:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/52/1003252/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/52/1003252/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/52/1003252/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/52/1003252/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/52/1003252/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/52/1003252/2"}}},"commit":{"parents":[{"commit":"f427fbaa343ef1e6a96c9a64706e2124975ad0a3","subject":"Update ansible_managed variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/f427fbaa343ef1e6a96c9a64706e2124975ad0a3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 10:53:30.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 10:54:31.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I45ec88253aca9bf98227d8c783349e3cf7e1dd3c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/af684e8f0308c872173d6e5403252e4d1dc66349"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/af684e8f0308c872173d6e5403252e4d1dc66349"}]},"branch":"refs/heads/master"},"e146401bc7784023c1c15ff8c680d29de2acdbcc":{"kind":"REWORK","_number":3,"created":"2026-09-06 23:50:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/52/1003252/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/52/1003252/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/52/1003252/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/52/1003252/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/52/1003252/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/52/1003252/3"}}},"commit":{"parents":[{"commit":"f427fbaa343ef1e6a96c9a64706e2124975ad0a3","subject":"Update ansible_managed variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/f427fbaa343ef1e6a96c9a64706e2124975ad0a3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 10:53:30.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 23:50:25.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: I45ec88253aca9bf98227d8c783349e3cf7e1dd3c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/e146401bc7784023c1c15ff8c680d29de2acdbcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/e146401bc7784023c1c15ff8c680d29de2acdbcc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
