)]}'
{"id":"openstack%2Fopenstack-ansible-os_nova~323377","triplet_id":"openstack%2Fopenstack-ansible-os_nova~master~Ic34011149f1355cfa48f4ee35537362e2d24e55d","project":"openstack/openstack-ansible-os_nova","branch":"master","topic":"bug/1587452","hashtags":[],"change_id":"Ic34011149f1355cfa48f4ee35537362e2d24e55d","subject":"Always execute Nova virt type detection","status":"MERGED","created":"2016-05-31 13:35:17.000000000","updated":"2016-06-03 01:12:08.000000000","submitted":"2016-06-03 01:12:08.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1930cabb73cefe91c9fc763a5f81966ecd308971","_number":323377,"virtual_id_number":323377,"owner":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2016-06-03 01:12:08.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":0,"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-06-02 21:24:37.000000000","_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":0,"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},{"value":2,"date":"2016-06-03 00:54:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":0,"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},{"value":1,"date":"2016-06-03 00:54:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":0,"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-06-02 09:47:09.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2016-06-02 21:24:37.000000000","updated_by":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"reviewer":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"state":"REVIEWER"},{"updated":"2016-06-03 00:54:23.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2016-06-03 01:12:08.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"b3a8eae1634af80c9ae5776f02bb5f9ab3e3fc70","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2016-05-31 13:35:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7172f022f293c6af1e2977e3f2a4c5807227792","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-31 14:09:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-openstack-ansible-os_nova-docs http://docs-draft.openstack.org/77/323377/1/check/gate-openstack-ansible-os_nova-docs/a728ca2//doc/build/html/ : SUCCESS in 1m 08s\n- gate-openstack-ansible-os_nova-linters http://logs.openstack.org/77/323377/1/check/gate-openstack-ansible-os_nova-linters/56e0dd7/ : SUCCESS in 1m 06s\n- gate-openstack-ansible-os_nova-ansible-func-ubuntu-trusty http://logs.openstack.org/77/323377/1/check/gate-openstack-ansible-os_nova-ansible-func-ubuntu-trusty/e9cc62f/ : FAILURE in 24m 02s","accounts_in_message":[],"_revision_number":1},{"id":"60768ab856eb3b17009508ac1029bf4d1e7a9cdb","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2016-06-01 22:04:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"182f642acaa4f852e61403e252d244f6ebf4d886","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-01 22:37:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-openstack-ansible-os_nova-docs http://docs-draft.openstack.org/77/323377/1/check/gate-openstack-ansible-os_nova-docs/9c8b306//doc/build/html/ : SUCCESS in 1m 10s\n- gate-openstack-ansible-os_nova-linters http://logs.openstack.org/77/323377/1/check/gate-openstack-ansible-os_nova-linters/b0bf320/ : SUCCESS in 2m 41s\n- gate-openstack-ansible-os_nova-ansible-func-ubuntu-trusty http://logs.openstack.org/77/323377/1/check/gate-openstack-ansible-os_nova-ansible-func-ubuntu-trusty/dcf1a3d/ : SUCCESS in 28m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2657402f6a01a2e006a5e7d8ab89b1dcc645738d","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2016-06-02 09:47:09.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"80b1ae2ddc0b78281108a760f5c55fd7e1e28c80","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2016-06-02 20:02:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b52ed8fac4d2c16bc2dfe674981d4ba0cfa69edc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-02 20:54:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-openstack-ansible-os_nova-docs http://docs-draft.openstack.org/77/323377/2/check/gate-openstack-ansible-os_nova-docs/cae89b9//doc/build/html/ : SUCCESS in 31s\n- gate-openstack-ansible-os_nova-linters http://logs.openstack.org/77/323377/2/check/gate-openstack-ansible-os_nova-linters/50c4b39/ : SUCCESS in 52s\n- gate-openstack-ansible-os_nova-ansible-func-ubuntu-trusty http://logs.openstack.org/77/323377/2/check/gate-openstack-ansible-os_nova-ansible-func-ubuntu-trusty/ff80f8c/ : SUCCESS in 23m 36s","accounts_in_message":[],"_revision_number":2},{"id":"ee0738c2464948603b4674e0d7b9fce98d5b646d","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2016-06-02 21:24:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a51ba24620fd755f4c97ddf5b7c73dd1c5301fcf","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2016-06-03 00:54:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"583938c2dd13a1922719207b04acf68476b1ce00","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-03 00:54:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"8cd4d6fce37280e7f8c6e518730765b04715bda3","date":"2016-06-03 01:12:08.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":2},{"id":"a98f46a0d9058ebdbe03a57c80c66eaa3e1686c3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-03 01:12:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-openstack-ansible-os_nova-docs http://docs-draft.openstack.org/77/323377/2/gate/gate-openstack-ansible-os_nova-docs/0ab8aa0//doc/build/html/ : SUCCESS in 54s\n- gate-openstack-ansible-os_nova-linters http://logs.openstack.org/77/323377/2/gate/gate-openstack-ansible-os_nova-linters/64d074d/ : SUCCESS in 41s\n- gate-openstack-ansible-os_nova-ansible-func-ubuntu-trusty http://logs.openstack.org/77/323377/2/gate/gate-openstack-ansible-os_nova-ansible-func-ubuntu-trusty/bbf2171/ : SUCCESS in 16m 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a8d1e412ef8af4ac345c5f926e37b741aa7a1adb","revisions":{"d9d1a6412cbb280a141147443f2a10c06d7daedb":{"kind":"REWORK","_number":1,"created":"2016-05-31 13:35:17.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/77/323377/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/77/323377/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/77/323377/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/77/323377/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/77/323377/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/77/323377/1"}}},"commit":{"parents":[{"commit":"1fc45e5465d3c6c78b02274344f25a438a0837a3","subject":"Merge \"Cleanup Nova console proxy git repos before updating it\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/1fc45e5465d3c6c78b02274344f25a438a0837a3"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2016-05-31 13:07:58.000000000","tz":-300},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2016-05-31 13:07:58.000000000","tz":-300},"subject":"Always execute Nova virt type detection","message":"Always execute Nova virt type detection\n\nThis fix executes the nova virt type detection on any playbook run\nand will eliminate the need to specify the nova-cpuinfo-read,\nnova-virt-type-set tags along with the intended tag like nova-config.\n\nChange-Id: Ic34011149f1355cfa48f4ee35537362e2d24e55d\nCloses-Bug: #1587452\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/d9d1a6412cbb280a141147443f2a10c06d7daedb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/d9d1a6412cbb280a141147443f2a10c06d7daedb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1fc45e5465d3c6c78b02274344f25a438a0837a3","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a8d1e412ef8af4ac345c5f926e37b741aa7a1adb":{"kind":"REWORK","_number":2,"created":"2016-06-02 20:02:17.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/77/323377/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/77/323377/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/77/323377/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/77/323377/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/77/323377/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/77/323377/2"}}},"commit":{"parents":[{"commit":"1fc45e5465d3c6c78b02274344f25a438a0837a3","subject":"Merge \"Cleanup Nova console proxy git repos before updating it\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/1fc45e5465d3c6c78b02274344f25a438a0837a3"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2016-05-31 13:07:58.000000000","tz":-300},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2016-06-02 20:02:14.000000000","tz":-300},"subject":"Always execute Nova virt type detection","message":"Always execute Nova virt type detection\n\nThis fix executes the nova virt type detection on any playbook run\nand will eliminate the need to specify the nova-cpuinfo-read,\nnova-virt-type-set tags along with the intended tag like nova-config.\n\nChange-Id: Ic34011149f1355cfa48f4ee35537362e2d24e55d\nCloses-Bug: #1587452\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/a8d1e412ef8af4ac345c5f926e37b741aa7a1adb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/a8d1e412ef8af4ac345c5f926e37b741aa7a1adb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1fc45e5465d3c6c78b02274344f25a438a0837a3","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
