)]}'
{"id":"openstack%2Fopenstack-ansible-os_nova~437328","triplet_id":"openstack%2Fopenstack-ansible-os_nova~master~I6c950cc31b66479dc980ecd5eb580d403dd27ae6","project":"openstack/openstack-ansible-os_nova","branch":"master","topic":"ravi_cap_worker_threads","hashtags":[],"change_id":"I6c950cc31b66479dc980ecd5eb580d403dd27ae6","subject":"Cap the number of worker threads","status":"MERGED","created":"2017-02-23 10:43:37.000000000","updated":"2017-02-23 16:12:08.000000000","submitted":"2017-02-23 16:12:08.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c0eca04a7666ee84195eb85fa5573da0eaed4390","_number":437328,"virtual_id_number":437328,"owner":{"_account_id":22111,"name":"Ravi Gummadi","email":"ravigumm@in.ibm.com","username":"ravigumm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2017-02-23 16:12:08.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2017-02-23 14:28:12.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-02-23 15:24:42.000000000","_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-02-23 15:24:42.000000000","_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"},{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-02-23 14:28:12.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2017-02-23 15:24:42.000000000","updated_by":{"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"},"reviewer":{"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"},"state":"REVIEWER"},{"updated":"2017-02-23 16:12:08.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"d3d42aba6e0abf49f9e3b9efa29244cc5eda4b30","author":{"_account_id":22111,"name":"Ravi Gummadi","email":"ravigumm@in.ibm.com","username":"ravigumm"},"date":"2017-02-23 10:43:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4da6f0ff3e227cf31c9b3f02bc117d799dcc4be","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-23 11:49:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-openstack-ansible-os_nova-docs-ubuntu-xenial http://docs-draft.openstack.org/28/437328/1/check/gate-openstack-ansible-os_nova-docs-ubuntu-xenial/48ae6c1//doc/build/html/ : SUCCESS in 35s\n- gate-openstack-ansible-os_nova-linters-ubuntu-xenial http://logs.openstack.org/28/437328/1/check/gate-openstack-ansible-os_nova-linters-ubuntu-xenial/172da8d/ : SUCCESS in 1m 05s\n- gate-openstack-ansible-os_nova-releasenotes http://docs-draft.openstack.org/28/437328/1/check/gate-openstack-ansible-os_nova-releasenotes/95bea55//releasenotes/build/html/ : SUCCESS in 1m 14s\n- gate-openstack-ansible-os_nova-ansible-func-centos-7-nv http://logs.openstack.org/28/437328/1/check/gate-openstack-ansible-os_nova-ansible-func-centos-7-nv/329a2f1/ : FAILURE in 1h 05m 16s (non-voting)\n- gate-openstack-ansible-os_nova-ansible-func-ubuntu-xenial http://logs.openstack.org/28/437328/1/check/gate-openstack-ansible-os_nova-ansible-func-ubuntu-xenial/d1b3950/ : SUCCESS in 31m 53s\n- gate-openstack-ansible-os_nova-ansible-func_lxd-ubuntu-xenial http://logs.openstack.org/28/437328/1/check/gate-openstack-ansible-os_nova-ansible-func_lxd-ubuntu-xenial/b4decfe/ : SUCCESS in 29m 32s\n- gate-openstack-ansible-os_nova-ansible-upgrade-ubuntu-xenial-nv http://logs.openstack.org/28/437328/1/check/gate-openstack-ansible-os_nova-ansible-upgrade-ubuntu-xenial-nv/e887bce/ : FAILURE in 30m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0dcb90fea7072e4f095f8ef463616cf18348c09","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2017-02-23 14:28:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"82547aff76984099cc4df1c1104382ae4900be5b","author":{"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"},"date":"2017-02-23 15:24:42.000000000","message":"Patch Set 1: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1cfd8924674b06141d3de38d73b7e23bb9259c20","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-23 15:24:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"a55aa963c4efe4d0f48870e26d6bcdb6fb0401d0","date":"2017-02-23 16:12:08.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"4fef7d5dbc04e0a01e6985ee8ee4e563f5715db6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-23 16:12:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-openstack-ansible-os_nova-docs-ubuntu-xenial http://docs-draft.openstack.org/28/437328/1/gate/gate-openstack-ansible-os_nova-docs-ubuntu-xenial/c54b079//doc/build/html/ : SUCCESS in 2m 18s\n- gate-openstack-ansible-os_nova-linters-ubuntu-xenial http://logs.openstack.org/28/437328/1/gate/gate-openstack-ansible-os_nova-linters-ubuntu-xenial/23218bb/ : SUCCESS in 3m 36s\n- gate-openstack-ansible-os_nova-releasenotes http://docs-draft.openstack.org/28/437328/1/gate/gate-openstack-ansible-os_nova-releasenotes/c49b1ab//releasenotes/build/html/ : SUCCESS in 1m 57s\n- gate-openstack-ansible-os_nova-ansible-func-ubuntu-xenial http://logs.openstack.org/28/437328/1/gate/gate-openstack-ansible-os_nova-ansible-func-ubuntu-xenial/eee6455/ : SUCCESS in 45m 39s\n- gate-openstack-ansible-os_nova-ansible-func_lxd-ubuntu-xenial http://logs.openstack.org/28/437328/1/gate/gate-openstack-ansible-os_nova-ansible-func_lxd-ubuntu-xenial/96b52fa/ : SUCCESS in 31m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f83296e07387affab9364b273c999f292e8c9905","revisions":{"f83296e07387affab9364b273c999f292e8c9905":{"kind":"REWORK","_number":1,"created":"2017-02-23 10:43:37.000000000","uploader":{"_account_id":22111,"name":"Ravi Gummadi","email":"ravigumm@in.ibm.com","username":"ravigumm"},"ref":"refs/changes/28/437328/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/28/437328/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/28/437328/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/28/437328/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/28/437328/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/28/437328/1"}}},"commit":{"parents":[{"commit":"5c99b101783ce2177d96d8641e832c5b5c79b615","subject":"Ordered service restarts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/5c99b101783ce2177d96d8641e832c5b5c79b615"}]}],"author":{"name":"Ravi Gummadi","email":"ravigumm@in.ibm.com","date":"2017-02-23 10:45:47.000000000","tz":-300},"committer":{"name":"Ravi Gummadi","email":"ravigumm@in.ibm.com","date":"2017-02-23 10:46:15.000000000","tz":-300},"subject":"Cap the number of worker threads","message":"Cap the number of worker threads\n\nUsers can configure the number of worker threads however when it\u0027s\nnot specified the calculated number of workers can get too large on\nhosts with a large number of CPUs.\n\nChange-Id: I6c950cc31b66479dc980ecd5eb580d403dd27ae6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/f83296e07387affab9364b273c999f292e8c9905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/f83296e07387affab9364b273c999f292e8c9905"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5c99b101783ce2177d96d8641e832c5b5c79b615","is_merged_in_target_branch":true,"change_id":"I29e17c09c6aa1b626aead8e4916cc89604a371d6","change_number":436165,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
