)]}'
{"id":"openstack%2Fopenstack-ansible-os_nova~494637","triplet_id":"openstack%2Fopenstack-ansible-os_nova~master~Ib0448b63b6ce7c7e5d48975b2bd9d78ef5c1218c","project":"openstack/openstack-ansible-os_nova","branch":"master","topic":"bug/1669436","hashtags":[],"change_id":"Ib0448b63b6ce7c7e5d48975b2bd9d78ef5c1218c","subject":"Fail if TLS + TCP are enabled for libvirtd","status":"MERGED","created":"2017-08-17 16:33:25.000000000","updated":"2017-08-18 14:47:48.000000000","submitted":"2017-08-17 21:16:07.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f1ed9cbb2145195aa88210d1384a4ebe9a622b9c","_number":494637,"virtual_id_number":494637,"owner":{"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-08-17 21:16:07.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-08-17 16:51:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2017-08-17 16:39:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-08-17 16:51:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-08-17 16:39:33.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2017-08-17 16:51:48.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2017-08-17 21:16:07.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"97f9e764894dcdcc9a746ebbd5608cef9b509e71","author":{"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"},"date":"2017-08-17 16:33:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f49aac2f62f688adf7d1f03dfd3d9abd74adf58","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2017-08-17 16:39:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8431621db837b9fd50e062c7b6f028145da0a17a","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2017-08-17 16:51:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c475127a3baf05e8cddc6131b6e5a215835de40","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-17 20:18:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-openstack-ansible-os_nova-docs-ubuntu-xenial http://docs-draft.openstack.org/37/494637/1/check/gate-openstack-ansible-os_nova-docs-ubuntu-xenial/ef69b40//doc/build/html/ : SUCCESS in 2m 11s\n- gate-openstack-ansible-os_nova-linters-ubuntu-xenial http://logs.openstack.org/37/494637/1/check/gate-openstack-ansible-os_nova-linters-ubuntu-xenial/dac1414/ : SUCCESS in 2m 36s\n- gate-openstack-ansible-os_nova-ansible-func-centos-7-nv http://logs.openstack.org/37/494637/1/check/gate-openstack-ansible-os_nova-ansible-func-centos-7-nv/70de3e3/ : FAILURE in 10m 41s (non-voting)\n- gate-openstack-ansible-os_nova-ansible-func-ubuntu-xenial http://logs.openstack.org/37/494637/1/check/gate-openstack-ansible-os_nova-ansible-func-ubuntu-xenial/31df554/ : SUCCESS in 1h 05m 26s\n- gate-openstack-ansible-os_nova-ansible-func-opensuse-423-nv http://logs.openstack.org/37/494637/1/check/gate-openstack-ansible-os_nova-ansible-func-opensuse-423-nv/c955740/ : SUCCESS in 1h 00m 34s (non-voting)\n- gate-openstack-ansible-os_nova-ansible-func_lxd-ubuntu-xenial http://logs.openstack.org/37/494637/1/check/gate-openstack-ansible-os_nova-ansible-func_lxd-ubuntu-xenial/90b1631/ : SUCCESS in 54m 04s\n- gate-openstack-ansible-os_nova-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/37/494637/1/check/gate-openstack-ansible-os_nova-ansible-upgrade-ubuntu-xenial/1aef64e/ : SUCCESS in 55m 34s","accounts_in_message":[],"_revision_number":1},{"id":"7530f8ce16dc38436f0a982be9147e279eb7de70","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-17 20:18:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"cca60014cb74cd42f7df51dc725f07acde217e70","date":"2017-08-17 21:16:07.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"699510e8c54dc5bf29fb36465b6a56b1dc13b377","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-17 21:16:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-openstack-ansible-os_nova-docs-ubuntu-xenial http://docs-draft.openstack.org/37/494637/1/gate/gate-openstack-ansible-os_nova-docs-ubuntu-xenial/f15141e//doc/build/html/ : SUCCESS in 1m 52s\n- gate-openstack-ansible-os_nova-linters-ubuntu-xenial http://logs.openstack.org/37/494637/1/gate/gate-openstack-ansible-os_nova-linters-ubuntu-xenial/fc27134/ : SUCCESS in 3m 11s\n- gate-openstack-ansible-os_nova-ansible-func-ubuntu-xenial http://logs.openstack.org/37/494637/1/gate/gate-openstack-ansible-os_nova-ansible-func-ubuntu-xenial/18c0b67/ : SUCCESS in 55m 55s\n- gate-openstack-ansible-os_nova-ansible-func_lxd-ubuntu-xenial http://logs.openstack.org/37/494637/1/gate/gate-openstack-ansible-os_nova-ansible-func_lxd-ubuntu-xenial/eaefae5/ : SUCCESS in 46m 14s\n- gate-openstack-ansible-os_nova-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/37/494637/1/gate/gate-openstack-ansible-os_nova-ansible-upgrade-ubuntu-xenial/addadc0/ : SUCCESS in 52m 49s","accounts_in_message":[],"_revision_number":1},{"id":"a1d17c20959d6fca287f94da714413732d4e2201","author":{"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"},"date":"2017-08-18 14:47:48.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/pike as commit 72a059b7900d5974b3053bc6c7c03c782c450012","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37a8b27287c75e7b370af8946f613b6457bbc060","revisions":{"37a8b27287c75e7b370af8946f613b6457bbc060":{"kind":"REWORK","_number":1,"created":"2017-08-17 16:33:25.000000000","uploader":{"_account_id":538,"name":"Major Hayden","email":"major@mhtx.net","username":"rackerhacker"},"ref":"refs/changes/37/494637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/37/494637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/37/494637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/37/494637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/37/494637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/37/494637/1"}}},"commit":{"parents":[{"commit":"f5e5279ab703ecfdcd13887935a659d70ea60888","subject":"Merge \"Correct Ansible deprecation warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/f5e5279ab703ecfdcd13887935a659d70ea60888"}]}],"author":{"name":"Major Hayden","email":"major@mhtx.net","date":"2017-08-17 16:32:07.000000000","tz":-300},"committer":{"name":"Major Hayden","email":"major@mhtx.net","date":"2017-08-17 16:33:13.000000000","tz":-300},"subject":"Fail if TLS + TCP are enabled for libvirtd","message":"Fail if TLS + TCP are enabled for libvirtd\n\nThis patch fails the playbook if TLS and TCP connectivity are both\nenabled. The role does not set certificates for libvirtd at this time\nand these configurations will cause libvirtd to fail.\n\nCloses-Bug: 1669436\nChange-Id: Ib0448b63b6ce7c7e5d48975b2bd9d78ef5c1218c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/37a8b27287c75e7b370af8946f613b6457bbc060"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/37a8b27287c75e7b370af8946f613b6457bbc060"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f5e5279ab703ecfdcd13887935a659d70ea60888","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
