)]}'
{"id":"openstack%2Fopenstack-ansible-os_nova~565958","triplet_id":"openstack%2Fopenstack-ansible-os_nova~master~I567081403e2aabe9e0747f40e5bf0976d0f949ff","project":"openstack/openstack-ansible-os_nova","branch":"master","topic":"bug/1732481","hashtags":[],"change_id":"I567081403e2aabe9e0747f40e5bf0976d0f949ff","subject":"Add option to define the qemu security driver","status":"ABANDONED","created":"2018-05-03 05:13:56.000000000","updated":"2019-05-27 00:23:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b9635c6dbcdb133ecc5605261857ec642c3bab49","_number":565958,"virtual_id_number":565958,"owner":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":-1,"date":"2019-05-07 04:41:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-05-25 13:14:52.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-05-07 04:41:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6aab1c07c092a541f8c555667a5f1bed4cf71a1","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2018-05-03 05:13:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71454a92b7ed225a942c2200799fbb8e5fc9b06b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-03 06:27:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/58/565958/1/check/build-openstack-sphinx-docs/04a2763/html/ : SUCCESS in 2m 21s\n- openstack-ansible-linters http://logs.openstack.org/58/565958/1/check/openstack-ansible-linters/9e2c1b5/ : SUCCESS in 5m 30s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/58/565958/1/check/openstack-ansible-functional-centos-7/d3ae872/ : FAILURE in 1h 05m 41s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/58/565958/1/check/openstack-ansible-functional-opensuse-423/05bf0ab/ : FAILURE in 47m 28s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/58/565958/1/check/openstack-ansible-functional-ubuntu-xenial/ea1eee6/ : FAILURE in 42m 59s\n- openstack-ansible-upgrade-ubuntu-xenial-nv http://logs.openstack.org/58/565958/1/check/openstack-ansible-upgrade-ubuntu-xenial-nv/8689ea2/ : FAILURE in 59m 53s (non-voting)\n- openstack-ansible-lxd-ubuntu-xenial http://logs.openstack.org/58/565958/1/check/openstack-ansible-lxd-ubuntu-xenial/7648d15/ : SUCCESS in 46m 57s\n- openstack-ansible-nova-ssl-nv http://logs.openstack.org/58/565958/1/check/openstack-ansible-nova-ssl-nv/5ab4fc4/ : FAILURE in 20m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad6ad11b6847ab06d19905b970d26f57d70eb0ab","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2018-05-03 14:54:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"42ac7958842d97aee07e0aeb107a8f85698274ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-03 15:53:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/58/565958/1/check/build-openstack-sphinx-docs/538862f/html/ : SUCCESS in 2m 30s\n- openstack-ansible-linters http://logs.openstack.org/58/565958/1/check/openstack-ansible-linters/37b911f/ : SUCCESS in 4m 17s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/58/565958/1/check/openstack-ansible-functional-centos-7/f6375ef/ : FAILURE in 9m 07s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/58/565958/1/check/openstack-ansible-functional-opensuse-423/a98b9a0/ : FAILURE in 55m 21s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/58/565958/1/check/openstack-ansible-functional-ubuntu-xenial/08af8dd/ : FAILURE in 43m 36s\n- openstack-ansible-upgrade-ubuntu-xenial-nv http://logs.openstack.org/58/565958/1/check/openstack-ansible-upgrade-ubuntu-xenial-nv/83f7304/ : FAILURE in 56m 22s (non-voting)\n- openstack-ansible-lxd-ubuntu-xenial http://logs.openstack.org/58/565958/1/check/openstack-ansible-lxd-ubuntu-xenial/b5cbba8/ : SUCCESS in 48m 48s\n- openstack-ansible-nova-ssl-nv http://logs.openstack.org/58/565958/1/check/openstack-ansible-nova-ssl-nv/44c74b6/ : FAILURE in 20m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b35c042a06e06e778f735e229887a6a7facb9fdf","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2018-05-15 03:48:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"244aa97fd6033aef822c65ed223995ca9097ac25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-15 04:56:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/58/565958/1/check/build-openstack-sphinx-docs/07162fa/html/ : SUCCESS in 2m 07s\n- openstack-ansible-linters http://logs.openstack.org/58/565958/1/check/openstack-ansible-linters/f005534/ : SUCCESS in 5m 07s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/58/565958/1/check/openstack-ansible-functional-centos-7/cb44bd4/ : FAILURE in 1h 03m 26s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/58/565958/1/check/openstack-ansible-functional-opensuse-423/5df59bb/ : FAILURE in 49m 24s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/58/565958/1/check/openstack-ansible-functional-ubuntu-xenial/068024e/ : FAILURE in 45m 05s\n- openstack-ansible-upgrade-ubuntu-xenial-nv http://logs.openstack.org/58/565958/1/check/openstack-ansible-upgrade-ubuntu-xenial-nv/8f4e606/ : FAILURE in 1h 00m 10s (non-voting)\n- openstack-ansible-lxd-ubuntu-xenial http://logs.openstack.org/58/565958/1/check/openstack-ansible-lxd-ubuntu-xenial/6376339/ : SUCCESS in 48m 34s\n- openstack-ansible-nova-ssl-nv http://logs.openstack.org/58/565958/1/check/openstack-ansible-nova-ssl-nv/d64a38b/ : FAILURE in 18m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"76f51de6a9a4c1c605eae0aaa5f5058fe7de07d6","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2018-05-25 13:14:52.000000000","message":"Patch Set 1:\n\nrecheck bug 1773248","accounts_in_message":[],"_revision_number":1},{"id":"371d13532be4dad07ccfcea4f931930cc6e9e76e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-25 14:44:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/58/565958/1/check/build-openstack-sphinx-docs/7df2527/html/ : SUCCESS in 1m 51s\n- openstack-ansible-linters http://logs.openstack.org/58/565958/1/check/openstack-ansible-linters/feb9486/ : SUCCESS in 3m 58s\n- openstack-ansible-functional-centos-7-nv http://logs.openstack.org/58/565958/1/check/openstack-ansible-functional-centos-7-nv/30c152c/ : FAILURE in 1h 24m 43s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/58/565958/1/check/openstack-ansible-functional-opensuse-423/c620e42/ : FAILURE in 48m 35s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/58/565958/1/check/openstack-ansible-functional-ubuntu-xenial/47e684e/ : FAILURE in 57m 39s\n- openstack-ansible-upgrade-ubuntu-xenial-nv http://logs.openstack.org/58/565958/1/check/openstack-ansible-upgrade-ubuntu-xenial-nv/6a5e44a/ : FAILURE in 13m 11s (non-voting)\n- openstack-ansible-lxd-ubuntu-xenial http://logs.openstack.org/58/565958/1/check/openstack-ansible-lxd-ubuntu-xenial/54921ef/ : SUCCESS in 56m 48s\n- openstack-ansible-nova-ssl-nv http://logs.openstack.org/58/565958/1/check/openstack-ansible-nova-ssl-nv/20829f0/ : FAILURE in 21m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"795abacd1379bc6c5169b2210acdc003ad639f27","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-05-07 03:12:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f43b7f4a6d2bddef3ed2f98f2d6f056aa5486a46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 04:41:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/58/565958/2/check/openstack-ansible-linters/55a8d4b/ : SUCCESS in 7m 53s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/58/565958/2/check/openstack-ansible-deploy-aio_metal-debian-stable/7d8cc86/ : SUCCESS in 52m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/58/565958/2/check/openstack-ansible-deploy-aio_metal-centos-7/fa7f187/ : SUCCESS in 1h 04m 07s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/58/565958/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/f052b9b/ : SUCCESS in 1h 06m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/58/565958/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/e9c60c2/ : SUCCESS in 57m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/58/565958/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/e2cbb83/ : FAILURE in 36m 27s\n- openstack-tox-docs http://logs.openstack.org/58/565958/2/check/openstack-tox-docs/5f15aa5/html/ : SUCCESS in 2m 32s\n- openstack-ansible-lxd-ubuntu-bionic http://logs.openstack.org/58/565958/2/check/openstack-ansible-lxd-ubuntu-bionic/2382bc2/ : FAILURE in 42m 34s","accounts_in_message":[],"_revision_number":2},{"id":"d67f37e88e0ba0eed2674808272a4ec954851564","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-05-27 00:23:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"70573513b304d239a95a97feefbc2766b1ad7807","revisions":{"b231083b8bc34b425f08edff6372e848ad4fee20":{"kind":"REWORK","_number":1,"created":"2018-05-03 05:13:56.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/58/565958/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/58/565958/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/58/565958/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/58/565958/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/58/565958/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/58/565958/1"}}},"commit":{"parents":[{"commit":"40f540ceab1d54115aab22fd1a0ae54700f4bfa1","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/40f540ceab1d54115aab22fd1a0ae54700f4bfa1"}]}],"author":{"name":"Kevin Carter","email":"kevin.carter@rackspace.com","date":"2018-05-03 05:12:33.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kevin.carter@rackspace.com","date":"2018-05-03 05:12:33.000000000","tz":-300},"subject":"Add option to define the qemu security driver","message":"Add option to define the qemu security driver\n\nIt is recommended to set the qemu security driver by default. This\nchange ensures that nova compute will define the security driver based\non a given OS deployment unless specifically defined by the deployer in\na user variable file.\n\nChange-Id: I567081403e2aabe9e0747f40e5bf0976d0f949ff\nCloses-Bug: #1732481\nSigned-off-by: Kevin Carter \u003ckevin.carter@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/b231083b8bc34b425f08edff6372e848ad4fee20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/b231083b8bc34b425f08edff6372e848ad4fee20"}]},"branch":"refs/heads/master"},"70573513b304d239a95a97feefbc2766b1ad7807":{"kind":"REWORK","_number":2,"created":"2019-05-07 03:12:36.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/58/565958/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/58/565958/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/58/565958/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/58/565958/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/58/565958/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/58/565958/2"}}},"commit":{"parents":[{"commit":"00e30808b31a7a135702a8e597187c1544bf7127","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/00e30808b31a7a135702a8e597187c1544bf7127"}]}],"author":{"name":"Kevin Carter","email":"kevin.carter@rackspace.com","date":"2018-05-03 05:12:33.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kevin@cloudnull.com","date":"2019-05-07 03:12:31.000000000","tz":-300},"subject":"Add option to define the qemu security driver","message":"Add option to define the qemu security driver\n\nIt is recommended to set the qemu security driver by default. This\nchange ensures that nova compute will define the security driver based\non a given OS deployment unless specifically defined by the deployer in\na user variable file.\n\nChange-Id: I567081403e2aabe9e0747f40e5bf0976d0f949ff\nCloses-Bug: #1732481\nSigned-off-by: Kevin Carter \u003ckevin.carter@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/70573513b304d239a95a97feefbc2766b1ad7807"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/70573513b304d239a95a97feefbc2766b1ad7807"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
