)]}'
{"id":"openstack%2Fopenstack-ansible-os_nova~666689","triplet_id":"openstack%2Fopenstack-ansible-os_nova~stable%2Frocky~I16ead452add6e25e91cc02c2bc063952ad7083fa","project":"openstack/openstack-ansible-os_nova","branch":"stable/rocky","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I16ead452add6e25e91cc02c2bc063952ad7083fa","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-20 20:02:16.000000000","updated":"2019-06-21 08:17:36.000000000","submitted":"2019-06-21 08:17:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666689-1561105056327-f027147f","meta_rev_id":"124082c8c82636681ee89fcaeccf7573d24ccfa2","_number":666689,"virtual_id_number":666689,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-06-21 08:17:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-06-20 20:07:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-21 07:12:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-21 07:12:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 20:07:56.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-21 07:12:31.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-21 08:17:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c83c8a3ac4c779d101190951e3289e865beada85","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-20 20:02:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01eede0ece6ecb6ad36ea655b36951bc6a4d2f5f","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-20 20:07:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cc62f5b7cf125b7bbecad9232131bdbef333e043","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 22:08:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/89/666689/1/check/openstack-ansible-linters/9ddf241/ : SUCCESS in 5m 42s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/89/666689/1/check/openstack-ansible-functional-centos-7/01b140b/ : SUCCESS in 57m 05s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/89/666689/1/check/openstack-ansible-functional-opensuse-423/e3b668e/ : FAILURE in 8m 34s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/89/666689/1/check/openstack-ansible-functional-opensuse-150/a1d84e8/ : FAILURE in 10m 56s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/89/666689/1/check/openstack-ansible-functional-ubuntu-bionic/c65d562/ : SUCCESS in 49m 32s (non-voting)\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/89/666689/1/check/openstack-ansible-functional-ubuntu-xenial/3cb8beb/ : TIMED_OUT in 1h 33m 38s\n- build-openstack-sphinx-docs http://logs.openstack.org/89/666689/1/check/build-openstack-sphinx-docs/428cf4f/html/ : SUCCESS in 3m 47s\n- openstack-ansible-lxd-ubuntu-xenial http://logs.openstack.org/89/666689/1/check/openstack-ansible-lxd-ubuntu-xenial/016af39/ : SUCCESS in 46m 46s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/89/666689/1/check/openstack-ansible-functional-distro_install-centos-7/81c88c9/ : SUCCESS in 50m 20s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/89/666689/1/check/openstack-ansible-functional-distro_install-opensuse-423/82c0b19/ : FAILURE in 8m 48s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-xenial http://logs.openstack.org/89/666689/1/check/openstack-ansible-functional-distro_install-ubuntu-xenial/2eb89d0/ : TIMED_OUT in 1h 52m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3cb1cf0ec88c124212053df93add76f1648daf7c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-20 22:09:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"185d173c57c0c1908780935136fad22067b72042","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 01:05:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/89/666689/1/check/openstack-ansible-linters/01ac212/ : SUCCESS in 7m 30s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/89/666689/1/check/openstack-ansible-functional-centos-7/30f4242/ : SUCCESS in 1h 06m 48s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/89/666689/1/check/openstack-ansible-functional-opensuse-423/a72e693/ : FAILURE in 9m 48s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/89/666689/1/check/openstack-ansible-functional-opensuse-150/ce7fec9/ : FAILURE in 9m 48s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/89/666689/1/check/openstack-ansible-functional-ubuntu-bionic/d9c9ce1/ : SUCCESS in 1h 06m 15s (non-voting)\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/89/666689/1/check/openstack-ansible-functional-ubuntu-xenial/a750438/ : SUCCESS in 48m 33s\n- build-openstack-sphinx-docs http://logs.openstack.org/89/666689/1/check/build-openstack-sphinx-docs/82b4e03/html/ : SUCCESS in 2m 51s\n- openstack-ansible-lxd-ubuntu-xenial http://logs.openstack.org/89/666689/1/check/openstack-ansible-lxd-ubuntu-xenial/ddd45da/ : SUCCESS in 45m 39s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/89/666689/1/check/openstack-ansible-functional-distro_install-centos-7/0569f6f/ : SUCCESS in 53m 48s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/89/666689/1/check/openstack-ansible-functional-distro_install-opensuse-423/a6e1366/ : FAILURE in 8m 13s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-xenial http://logs.openstack.org/89/666689/1/check/openstack-ansible-functional-distro_install-ubuntu-xenial/72b047f/ : TIMED_OUT in 1h 37m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5514148f1117ce3d74f206cb1ba6a1c277041899","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-21 07:12:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a9325209905a45a435957cfc59905b1348eebc67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 07:12:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20b008866e987d2e299871e24bece500e3a5a85e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 08:17:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d1cbb0fbe6bcd0ad734c29e5ec458888fdd4461c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 08:17:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/89/666689/1/gate/openstack-ansible-linters/4e3e3db/ : SUCCESS in 5m 54s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/89/666689/1/gate/openstack-ansible-functional-centos-7/f3bfd8b/ : SUCCESS in 58m 44s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/89/666689/1/gate/openstack-ansible-functional-opensuse-423/09dfa20/ : FAILURE in 8m 03s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/89/666689/1/gate/openstack-ansible-functional-ubuntu-bionic/6fc573f/ : SUCCESS in 49m 21s (non-voting)\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/89/666689/1/gate/openstack-ansible-functional-ubuntu-xenial/373d69e/ : SUCCESS in 50m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/89/666689/1/gate/build-openstack-sphinx-docs/6903af8/html/ : SUCCESS in 3m 12s\n- openstack-ansible-lxd-ubuntu-xenial http://logs.openstack.org/89/666689/1/gate/openstack-ansible-lxd-ubuntu-xenial/8c385d7/ : SUCCESS in 52m 37s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/89/666689/1/gate/openstack-ansible-functional-distro_install-centos-7/bec2500/ : SUCCESS in 52m 07s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/89/666689/1/gate/openstack-ansible-functional-distro_install-opensuse-423/2751bad/ : FAILURE in 8m 11s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f81cb507a082f91f5f656e29be8829a70405a306","revisions":{"f81cb507a082f91f5f656e29be8829a70405a306":{"kind":"REWORK","_number":1,"created":"2019-06-20 20:02:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/666689/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/89/666689/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/89/666689/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/89/666689/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/89/666689/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/89/666689/1"}}},"commit":{"parents":[{"commit":"baaf97d1bab78ca55f1eb44c2f7bee895fc2b041","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/baaf97d1bab78ca55f1eb44c2f7bee895fc2b041"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-20 20:02:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-20 20:02:14.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I16ead452add6e25e91cc02c2bc063952ad7083fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/f81cb507a082f91f5f656e29be8829a70405a306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/f81cb507a082f91f5f656e29be8829a70405a306"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
