)]}'
{"id":"openstack%2Fopenstack-ansible-os_nova~743194","triplet_id":"openstack%2Fopenstack-ansible-os_nova~stable%2Fussuri~I3f284e6d32f4aa614f311d3e4e754830be2f11a6","project":"openstack/openstack-ansible-os_nova","branch":"stable/ussuri","topic":"avoid-rebuilding-venv-with-each-run-stable/ussuri","hashtags":[],"change_id":"I3f284e6d32f4aa614f311d3e4e754830be2f11a6","subject":"Avoid rebuilding the venv with every run","status":"MERGED","created":"2020-07-27 11:08:00.000000000","updated":"2020-08-11 13:03:41.000000000","submitted":"2020-08-11 13:02:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743194-1597150926935-de949bf5","meta_rev_id":"18bc1c3ffc56a32fbd86d7f8a2205522ebb83108","_number":743194,"virtual_id_number":743194,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30660,"name":"Jan Marquardt","email":"jm@artfiles.de","username":"janno"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-11 13:02:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":30660,"name":"Jan Marquardt","email":"jm@artfiles.de","username":"janno"},{"value":2,"date":"2020-08-06 19:40:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":30660,"name":"Jan Marquardt","email":"jm@artfiles.de","username":"janno"},{"value":1,"date":"2020-08-11 10:40:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30660,"name":"Jan Marquardt","email":"jm@artfiles.de","username":"janno"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":30660,"name":"Jan Marquardt","email":"jm@artfiles.de","username":"janno"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-27 11:08:00.000000000","updated_by":{"_account_id":30660,"name":"Jan Marquardt","email":"jm@artfiles.de","username":"janno"},"reviewer":{"_account_id":30660,"name":"Jan Marquardt","email":"jm@artfiles.de","username":"janno"},"state":"REVIEWER"},{"updated":"2020-08-11 10:40:45.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-08-11 13:02:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f12878f089ddb971632b72ba693ee1d83c183077","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-27 11:08:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a82ca8daa3261683488cafb4112a4ff0487e0caa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 13:17:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/01a5d6cd64064100abed07e747bcb2a7 : SUCCESS in 10m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6d1cd70377824f59a0c31dd00959fef9 : SUCCESS in 1h 13m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ff66b4cde2574571b2c5304c769e656d : SUCCESS in 1h 35m 29s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4b8208fb7eaf4cc8b3c1ad1690bfeea9 : SUCCESS in 1h 36m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd7ae2c6b0574ae384a9d926f8aec206 : SUCCESS in 1h 12m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6498feb0b2584b0e8e5360dbf3e7667c : SUCCESS in 1h 10m 37s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/96840c1b3c5b44ca84e47fa4a67c646e : FAILURE in 18m 16s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43074db248e141ecb2bf3dda417f861e : FAILURE in 1h 41m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/894303eb2e304dabbef8fb51e5f27d07 : SUCCESS in 1h 03m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f89e91dfc12c4fbcb309a347cd16bce7 : SUCCESS in 1h 02m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96fc45b9da0541a7af13ac9024997925 : SUCCESS in 1h 02m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9182841a978452fb8a5f0d1f56d7197 : SUCCESS in 8m 08s","accounts_in_message":[],"_revision_number":1},{"id":"10439aec06ac4cb43449444325239806a0a04f80","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-06 19:39:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"106ff0847b3f6f43c91a846be90a7681f53337ec","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-06 19:40:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"51d51bed24e3b4248a4fdf74d5744bedfb2d4073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 22:45:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5677b6e0105d46c5ad3e5c067d51ed58 : SUCCESS in 6m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/aa7fe645147e40acb10612cdb5b149e2 : SUCCESS in 1h 11m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/de4a0051cc82444c8852f9eb071d6291 : SUCCESS in 1h 32m 11s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0615a4b8d9534155894e51dfcac7987d : SUCCESS in 1h 27m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3371be6cdf3c4dcb860f9d26fd27e01b : SUCCESS in 1h 10m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b37f03ac65d4ba0a3ea809b099698df : SUCCESS in 1h 10m 14s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b33d093e3eb749b6ac1663f313bbc1ba : SUCCESS in 2h 54m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7fc108b4274e42238eb358d0b140e271 : SUCCESS in 1h 47m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4fbafee727154a5a8943b6e997632478 : SUCCESS in 1h 05m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/008023db816f4c90ace7c2fed894f2e8 : SUCCESS in 55m 33s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/715d8f572524405e80968485bde309bf : SUCCESS in 52m 42s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/912e1764d1f34a2d8302ceb6240778c1 : SUCCESS in 53m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48b799a442054fe3842aeb36d56423bc : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":2},{"id":"5ce961f9568d555adccbd36da8736dfb824df287","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-11 10:40:11.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7e80ee48da36d78be59f02a7194dba6e224ee78b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 10:40:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0f846f5b9300733c23bf303f48c1a342ee2e66c4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-11 10:40:40.000000000","message":"Removed Workflow+1 by Dmitriy Rabotyagov (noonedeadpunk) \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"c2f15048fd9e61966e06f0e8d9cdd1cfbbcb9ae9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-11 10:40:45.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ae10ab8d980c635189a9d0ce7ca5a38e72b97d1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 13:02:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bf8ea6ceef9f452b859c539d03ae3b38 : SUCCESS in 7m 25s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/010b9855191f4349afe5c76bf69398df : SUCCESS in 1h 01m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d3dab599154e4a248562c6a93d22cde4 : SUCCESS in 1h 18m 22s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bd45fc3e7df64e93bb8c86f7f6183c8c : SUCCESS in 54m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9adc55aad2404d368e10a304d7f6fb17 : SUCCESS in 1h 00m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39932ccc3d0e4e2192cda18841cd7f17 : SUCCESS in 2h 06m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a512d4b848f04d78acf9083b8b2b6a5f : SUCCESS in 1h 00m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9fa4c35c8f9b457d9a0732321c1ecfae : SUCCESS in 54m 45s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d547c6e52b99445db12871d8dc43a758 : SUCCESS in 53m 04s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4ec9077250a4ea2bf4497adb5af327b : SUCCESS in 47m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0572d0280f364e688d087c7d2c4dedcb : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":2},{"id":"4cb27f2a27b6ff7e4713b431dd9f25be39272e0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 13:02:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"af127a1fd3ece2e5695171bb481658d01bc9d83a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 13:03:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d140cc384d045e98472001230b06a97 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7d46cb48d9f64ee4951553e7dd62e6ee37e70aea","revisions":{"c09a7aafe76ad25ea1e599edd25847007c734b6f":{"kind":"REWORK","_number":1,"created":"2020-07-27 11:08:00.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/94/743194/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/94/743194/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/94/743194/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/94/743194/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/94/743194/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/94/743194/1"}}},"commit":{"parents":[{"commit":"aeae89242e370737f39031be7c785dc2597e2875","subject":"Add Centos-8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/aeae89242e370737f39031be7c785dc2597e2875"}]}],"author":{"name":"Jan Marquardt","email":"jm@artfiles.de","date":"2020-07-22 08:19:06.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-27 11:08:00.000000000","tz":0},"subject":"Avoid rebuilding the venv with every run","message":"Avoid rebuilding the venv with every run\n\nUntil now the necessary pip packages differ between nova-api and\ncompute hosts, when novnc is used, because \u0027websockify\u0027 was only\nadded to nova-api containers. As a result, the venv is rebuilt twice\nfor each run. With this change \u0027websockify\u0027 is always added to\nthe needed pip packages.\n\nChange-Id: I3f284e6d32f4aa614f311d3e4e754830be2f11a6\n(cherry picked from commit 40bf388734a15d348100417d66bdc3de8fcd799d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/c09a7aafe76ad25ea1e599edd25847007c734b6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/c09a7aafe76ad25ea1e599edd25847007c734b6f"}]},"branch":"refs/heads/stable/ussuri"},"7d46cb48d9f64ee4951553e7dd62e6ee37e70aea":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-06 19:39:26.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/94/743194/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/94/743194/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/94/743194/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/94/743194/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/94/743194/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/94/743194/2"}}},"commit":{"parents":[{"commit":"dbe45e0cf6197b56f9c7bd1e23cfb0d997c160b6","subject":"Merge \"Use nova_management_address as a default VNC bind address\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/dbe45e0cf6197b56f9c7bd1e23cfb0d997c160b6"}]}],"author":{"name":"Jan Marquardt","email":"jm@artfiles.de","date":"2020-07-22 08:19:06.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-08-06 19:39:26.000000000","tz":0},"subject":"Avoid rebuilding the venv with every run","message":"Avoid rebuilding the venv with every run\n\nUntil now the necessary pip packages differ between nova-api and\ncompute hosts, when novnc is used, because \u0027websockify\u0027 was only\nadded to nova-api containers. As a result, the venv is rebuilt twice\nfor each run. With this change \u0027websockify\u0027 is always added to\nthe needed pip packages.\n\nChange-Id: I3f284e6d32f4aa614f311d3e4e754830be2f11a6\n(cherry picked from commit 40bf388734a15d348100417d66bdc3de8fcd799d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/7d46cb48d9f64ee4951553e7dd62e6ee37e70aea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/7d46cb48d9f64ee4951553e7dd62e6ee37e70aea"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
