)]}'
{"id":"openstack%2Fopenstack-ansible-os_nova~758411","triplet_id":"openstack%2Fopenstack-ansible-os_nova~stable%2Fussuri~I4953ee9de704c90dac67a70f9296062e5a071a77","project":"openstack/openstack-ansible-os_nova","branch":"stable/ussuri","topic":"enable-notifications-designate-stable/ussuri","hashtags":[],"change_id":"I4953ee9de704c90dac67a70f9296062e5a071a77","subject":"Enable notifications when Designate is enabled","status":"MERGED","created":"2020-10-15 12:09:15.000000000","updated":"2020-10-16 17:15:36.000000000","submitted":"2020-10-16 17:14:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758411-1602868457373-577aa814","meta_rev_id":"8e27de5605313a53ac23c6350408a0f8085dfc00","_number":758411,"virtual_id_number":758411,"owner":{"_account_id":819,"name":"Adrien Cunin","email":"adrien@adriencunin.fr","username":"adri2000"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-16 17:14:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":2,"date":"2020-10-15 12:20:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2020-10-15 18:58:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":1,"date":"2020-10-15 18:58:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-15 12:20:56.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2020-10-15 18:58:31.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-10-16 17:14:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"be1e4865ea09f4f7c0c44248a99d9f9321edfa85","author":{"_account_id":819,"name":"Adrien Cunin","email":"adrien@adriencunin.fr","username":"adri2000"},"date":"2020-10-15 12:09:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ce4f4795e3f0d755b082bfe222ef1b6b1721243b","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2020-10-15 12:20:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"617786c761969472afa9070d3ba5f0f65f9844a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 14:50:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f24e89d566ee413c861516ce7ddcb16d : SUCCESS in 7m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6ee0d6e225b440a7b2ee775eb69dc4a8 : SUCCESS in 1h 04m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/59af385ace8d4b09899a9adf1a19d723 : SUCCESS in 1h 31m 40s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c499fb46c13745eaafc6afff7995c1da : SUCCESS in 1h 17m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/771138ee810c48ec9a627172418b796d : SUCCESS in 55m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e8903c771a34a7a8b0b4bfcf9c06cef : FAILURE in 2h 32m 31s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3181f7c3a7d84dadabce1d828d01c686 : FAILURE in 2h 29m 58s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f635f4903a884ca198e97df9ac37faca : SUCCESS in 1h 56m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d7e4adf76a174ac4bd7517707556b0c7 : SUCCESS in 1h 02m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1cc228b05aae422f8a35b574aa89d324 : SUCCESS in 55m 18s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a9951ab5d634c16b942e319e26f3b5d : SUCCESS in 56m 32s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa9c045a6d9d45db94ec1e4d9681bd5e : SUCCESS in 55m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83dc00c0f4144692b18738e3a3cfb263 : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":1},{"id":"3e3b8b1e773db62e43686afdb06b4855566805b4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-15 18:58:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6229365418ed339406cb9e843e9451029df10014","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 21:47:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7ce92298f9f64f2db49bb0a6e36f96d4 : SUCCESS in 8m 02s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1c53059aedbc484a9087218391024a91 : FAILURE in 57m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e4c8f48a9e7240b495b7ac4f69c3ca06 : SUCCESS in 1h 31m 42s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/453b4052196f4b1eb9530df29414677b : SUCCESS in 1h 16m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da991b4be0eb4f1f889dc7eaf9d43a06 : SUCCESS in 1h 08m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd14ae42e5be4a248050423850e37322 : SUCCESS in 1h 07m 06s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1872ac712c0e4934b54934bde412c2b4 : SUCCESS in 2h 37m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a10e7deecabb4f99b2d4469fd9c11442 : SUCCESS in 1h 44m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1e2ebe2f104d4601a16f75f061b139c1 : SUCCESS in 1h 26m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/996b0635ced74adca9ee5eb612a66099 : SUCCESS in 1h 04m 02s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8219d5af83064c83a5cb23ba5587a8f0 : SUCCESS in 59m 50s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7642b43e23d8438fb6fdc7fabf02d043 : SUCCESS in 57m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10e3f22694a245ca99548df071d1757b : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"532509d3987862041ecfc8353029ca907e65824e","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2020-10-16 07:01:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5ee240244d7a150551904c0fc0f08581c0790ede","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 10:16:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/92f9af94e8cd4291b91a0b28992af3cb : SUCCESS in 8m 37s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4bd5ad6d705a4bf193eb97e0895fc419 : SUCCESS in 1h 00m 57s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/01b1bc69bf8845f7b0aeb97ddd146836 : SUCCESS in 1h 22m 38s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/47e57f1c87204fd79e529300ec9da395 : SUCCESS in 1h 23m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e2291231c3d4621b8673c1623cb45a7 : SUCCESS in 1h 09m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84b75d5aedb3451abfdf5ccc54cae12c : SUCCESS in 1h 03m 04s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c1e67e3101194dc9911ddf99a8a8bac6 : SUCCESS in 2h 57m 20s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49277425ac99486ab77100de559401f9 : SUCCESS in 2h 02m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cc5457f673fc431281216b6c1494b77e : SUCCESS in 1h 09m 55s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a1a5a8f0aa224b7a98289723e3dbfb53 : SUCCESS in 1h 03m 25s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e61fd0ce1e3a4d348fb7c378dd9b8de8 : SUCCESS in 1h 02m 24s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53b843f7d068426e8cc48e7f103fc010 : SUCCESS in 52m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42a4516e50354b50a9d2641e8a90fbe3 : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":1},{"id":"378fa555eab970657598d5eeae00bb77c29ec949","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 10:16:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7594a7c82fe4e83c4b3406af9a8b7e33c6f7e7c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 13:05:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/29b5c02620b54e0eb4a521417493aaa8 : SUCCESS in 8m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0343b151dda64f479609083029b4c462 : SUCCESS in 2h 36m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/89b7c9ad64fc404bbc7171466e9d4400 : SUCCESS in 1h 18m 58s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b90c86a917144fc0bd3ebf8948a1b31c : SUCCESS in 2h 06m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ebc09d2459044658b1bfb9b838be3a3 : SUCCESS in 2h 43m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db3d676db086406da57ad988ce3ec201 : FAILURE in 35m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/398a59e1aace4fa69523cfd5369ee2e7 : SUCCESS in 2h 45m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2e64444cd0a645edb2ad395118d4ada3 : SUCCESS in 2h 31m 00s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f55d2c3a5d6440fb75af896b0c98cc6 : SUCCESS in 1h 26m 33s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85fa949630034227afcba476e6a19620 : FAILURE in 2h 31m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5343eef22dac4e21bd900e5740a452c7 : SUCCESS in 11m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e07dde03ba5d79e79bd0cfc8358cfdfc52e9de2f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-16 13:23:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"560d206e6c29f7c93effe86166ab7383f188faca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 15:41:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/12ff39a18c41428d877d129515c781f7 : SUCCESS in 8m 25s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/36c581763ba145339ecac84254291aec : SUCCESS in 1h 11m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dba9918842564fbc8c49e7a2303df8a8 : SUCCESS in 1h 00m 22s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/132566673b8c4018a4ac017d0b7e72bb : SUCCESS in 1h 18m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0eae019a78cb42e19ae91cb468d89287 : SUCCESS in 2h 11m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46dfa0ad39544bccad3af0745572481c : SUCCESS in 1h 14m 28s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5d1efce488dc4b80b5bc1c979e15b211 : SUCCESS in 1h 43m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66e76e4844e149fea86d1460a81fb557 : SUCCESS in 1h 51m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f965f8f1988045efa85eb76e4bc8a83b : SUCCESS in 47m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a045ea0c729d460ebe1e862204a9ce4e : SUCCESS in 1h 03m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f43178b58ac443a7bae9088d4c346e01 : SUCCESS in 47m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ee890e52aa34742bc62cdc8695dfacb : SUCCESS in 59m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2760318a16624146806bddecc9d259ba : SUCCESS in 7m 24s","accounts_in_message":[],"_revision_number":1},{"id":"dd0d6d7ab3e9a273178bde94dd24c535556d3d48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 15:41:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5529ad668aa616be9da38266c86dec2586d84b68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 17:14:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f2112af6cdbbb39d0d176cf7985aca92a56da833","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 17:14:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d73846f62a884ca7b1ec9ccd6e228d96 : SUCCESS in 13m 22s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/974a21d052b44e16b081f920c30be0d9 : SUCCESS in 1h 14m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e279797486464127b779adf13cae6fd8 : SUCCESS in 1h 26m 58s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/49917dcd76c244559ced831f7d06c2f4 : SUCCESS in 1h 27m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c894899630c14c26b23bf611ffd630e4 : SUCCESS in 1h 20m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/787d6ff3924a479aac52e05b8d3eaaf7 : SUCCESS in 1h 10m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8275335bcc8c46bab42e770b4a5a758f : SUCCESS in 53m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7bdeed71b9cb4847869a883b7922e0fd : SUCCESS in 1h 06m 34s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5738c9aa6e184fbfb1281ee56df2c119 : SUCCESS in 1h 03m 15s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99a12d1c2a0044d29ae804d727128359 : SUCCESS in 58m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf6b8ef2a092473f9e96ee84c0bd7ccd : SUCCESS in 8m 29s","accounts_in_message":[],"_revision_number":1},{"id":"4fb380200b26b133ce4dbf0e6aac7132504b4b41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 17:15:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/547bafaeef4c488e8abdf9d5b2160913 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a90c631d5725f45ad45f2d6527f0fc0ad98d8559","revisions":{"a90c631d5725f45ad45f2d6527f0fc0ad98d8559":{"kind":"REWORK","_number":1,"created":"2020-10-15 12:09:15.000000000","uploader":{"_account_id":819,"name":"Adrien Cunin","email":"adrien@adriencunin.fr","username":"adri2000"},"ref":"refs/changes/11/758411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/11/758411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/11/758411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/11/758411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/11/758411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/11/758411/1"}}},"commit":{"parents":[{"commit":"004dc25f1a920d3acf58f74c3c21f91c8da6a91d","subject":"Merge \"Simplify scheduler filter additions\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/004dc25f1a920d3acf58f74c3c21f91c8da6a91d"}]}],"author":{"name":"Adrien Cunin","email":"adrien.cunin@springernature.com","date":"2020-10-13 15:55:31.000000000","tz":120},"committer":{"name":"Adrien Cunin","email":"adrien@adriencunin.fr","date":"2020-10-15 12:09:15.000000000","tz":0},"subject":"Enable notifications when Designate is enabled","message":"Enable notifications when Designate is enabled\n\nChange-Id: I4953ee9de704c90dac67a70f9296062e5a071a77\n(cherry picked from commit 881620bd6476d008b218721b70807deca451b8c6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/a90c631d5725f45ad45f2d6527f0fc0ad98d8559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/a90c631d5725f45ad45f2d6527f0fc0ad98d8559"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
