)]}'
{"id":"openstack%2Fopenstack-ansible-os_nova~941388","triplet_id":"openstack%2Fopenstack-ansible-os_nova~master~I335f0f1bcdb5e5564ce3f82f44eec7d8c6ab4e0e","project":"openstack/openstack-ansible-os_nova","branch":"master","topic":"osa/fqcn","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-03-03 15:47:05.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-03-03 15:43:54.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I335f0f1bcdb5e5564ce3f82f44eec7d8c6ab4e0e","subject":"Auto-fix usage of modules via FQCN","status":"MERGED","created":"2025-02-12 13:07:22.000000000","updated":"2025-03-03 17:33:13.000000000","submitted":"2025-03-03 17:32:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941388-osa/fqcn","meta_rev_id":"cbb6a967fc2aceb958bad0c4339cc9ed7f13da43","_number":941388,"virtual_id_number":941388,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-03 17:32:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2025-03-03 14:30:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-28 11:43:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-03-03 15:43:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-03-03 15:43:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-12 15:17:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-28 11:43:36.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-02-28 11:43:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-03-03 14:30:26.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"65e9e6fee94b40e26bf2b69cddbd419e44ba9c82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-12 13:07:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbfef0b9b812bd12bfdff3d70635392a70456e59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 15:17:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/69b9baf051a84c8ea377edf51b455f01\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d86ff743a1f64511b6d9d46eb71b563f : FAILURE in 5m 08s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3cdec42717e94d94b25d2c5554e168bf : SUCCESS in 1h 14m 08s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4cf223242c7b40fa9695156abe879885 : SUCCESS in 1h 27m 36s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/91387125413e4580a41908449838f7bb : SUCCESS in 1h 17m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66fef1b8fb4a4d9c965dee5b73783813 : SUCCESS in 1h 21m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1a49ec30454465d8ae2362958955ae1 : SUCCESS in 1h 23m 12s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2d6527bc06b249c7a847710b252aa512 : SUCCESS in 1h 26m 40s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/608fec17b4d24d63916591c931e9400d : SUCCESS in 1h 19m 17s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a117bf961fd44478d12cc7710736726 : SUCCESS in 1h 23m 49s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/67ebe9d0d23240cf802d2eafb4668703 : SUCCESS in 1h 49m 38s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/01e5d132444d4de3bba15578740451c4 : SUCCESS in 2h 00m 36s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3770f4a1c0144cb3b5f24fb7c1917e46 : SUCCESS in 1h 52m 25s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4015b88830514bb890dfa4e7755859ce : SUCCESS in 1h 47m 39s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fa7528be76384f00bbe74f32ba85fc15 : SUCCESS in 1h 55m 11s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a0c3771d8934d7380328b5acd3ae2fa : SUCCESS in 1h 38m 03s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2d985ad09af345dfb41da372591b909a : SUCCESS in 55m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0c6d960dc5f47d1a8124d3be79d74b4 : SUCCESS in 59m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbdf3271034443a5a574e68c034f9e6f : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5f8ef62d6ce8af44f2a4f314eb1194c521ccc04e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-18 12:00:46.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9d0d8fff121dfba5cef24adb9092ebc2cf3f649b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 14:40:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/46a5644627fd438c8efd121c9ef934b3\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49ab853398c343e485d3fd09f5b34978 : FAILURE in 7m 59s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/785da06d929042059b198148b0d6ced9 : SUCCESS in 1h 24m 14s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6f16210207043519d7104e80f589fd9 : SUCCESS in 1h 21m 53s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/83a27d9bfbcb4bbd874da5aadc20f9f6 : SUCCESS in 1h 17m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ae4cdde53714241897378e785f243a7 : SUCCESS in 1h 32m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b30fd09962e430a96bc60dcfff5c011 : SUCCESS in 46m 18s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/624f2f90c7a54554b6b9cee9b9fdc82e : SUCCESS in 1h 24m 53s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37f43eabf15f467282f0cbe8e416f332 : SUCCESS in 1h 17m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7c31da7ad9842e6b5a49dcf95a8c1f9 : SUCCESS in 1h 18m 24s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/51767e98c1854db98b53df350e7c8b62 : SUCCESS in 2h 30m 16s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9e2bf40be3dd465a836289f12f56a37d : SUCCESS in 2h 06m 43s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a0ba1bcabb54214bad5479ce62bc26b : SUCCESS in 1h 59m 28s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/daccd90bccdb47fe8461dcbf5416de48 : SUCCESS in 2h 16m 52s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9ace3b2a140d49f4a747df309a7aee67 : SUCCESS in 1h 50m 15s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02598378a5754c2b829cb14d359a9a86 : SUCCESS in 1h 50m 13s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0e7508b4cfca440dacc967b5b2672b88 : SUCCESS in 57m 41s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94517497db1f4f7983d51f2ae94c827f : SUCCESS in 1h 18m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a94730bbf544dbcb26812103c8734b8 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":2},{"id":"33d4bb900c909a8e091b6635f9dd08ba786300a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-28 08:38:53.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2b05f0f47907ead61cd82aa114bb79d03dcbd5a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 11:40:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e1eef415a6444968b441f488ffe51f1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87e52a5c8caf4de29e298011c8fd4878 : SUCCESS in 6m 28s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/75be44e74d1d456ba792ea9d5f486e3c : SUCCESS in 1h 23m 02s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/343c5192b17c42ca9f08e97bba0ddbdf : SUCCESS in 1h 32m 14s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bd5f985b42174d92be63cc92be23f4e7 : SUCCESS in 42m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28e3c04eebfb4b55b4c112027669a243 : SUCCESS in 1h 12m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7c532a310da478fb44df73880bec099 : SUCCESS in 1h 33m 02s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2017e4e72be345ffa4d36f560f5e8d63 : SUCCESS in 51m 12s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35266f6f52f647a4b7a9e72af037297a : SUCCESS in 1h 26m 16s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/928f6a6622594b488841f051f704076e : SUCCESS in 1h 20m 49s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/11d8b385a5c943cdbba5d7735868c409 : SUCCESS in 1h 53m 15s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/df278b9a9ba54617817181f79c7338b2 : SUCCESS in 1h 56m 59s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/207de550442240cdafac568e74483d27 : SUCCESS in 1h 46m 34s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa2edefed7794c04934ff96e8b59edd3 : SUCCESS in 1h 47m 56s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/180882bae4c74802a568cfee4f8b5e02 : SUCCESS in 2h 21m 38s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a6ec00150ea4c6db652d0cf790f35d9 : SUCCESS in 1h 53m 51s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/998b910eb884478e85134e7c4dc484b0 : SUCCESS in 1h 00m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4efec4278ddf4f9397d1fa93a7e97c42 : SUCCESS in 1h 02m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58fd05dcb11d4fa89b3432d93fc604eb : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":3},{"id":"db2d0973695f109d1989de795060e17f234bc154","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-28 11:43:36.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"cf2d744ae2284cf1d9c3a2dfafdd33b502b322ed","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2025-03-03 14:30:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b22e71af9cb8abf764b9eb0a1ee626c412dda1d6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-03-03 15:43:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2d8bbf211e367a834288305d9fcf82da295a2f5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 15:45:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a1e9909c39449d48768bed4ae5fec47bf4570eb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 17:32:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe778a290681425298fb9c4b20869bde\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd81b472e5db4a7a95b8935d7de5eff3 : SUCCESS in 7m 42s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e2f8d0995faa4c73b2ef9f428fe795da : SUCCESS in 1h 04m 47s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1e08535584d644758a380bd8baa32aac : SUCCESS in 1h 15m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58e7a905be3c441597b190a4062e778e : SUCCESS in 1h 26m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47411ad93dbe4b1187afbc87bfd759d8 : SUCCESS in 1h 15m 53s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4bd62afeedd14bf3bfbfa3a0cf237816 : SUCCESS in 1h 16m 54s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3aae8fe07ccc43fa8b33b388e0dc1b86 : SUCCESS in 1h 15m 11s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eaaa5d2c90e046228dac6b649daa531e : SUCCESS in 1h 38m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c775896c5e8432dbb446a9fe09d0861 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":3},{"id":"74227ad2e4ef29c0c1836e6ec4f0b21fc6188d3d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 17:32:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"cbb6a967fc2aceb958bad0c4339cc9ed7f13da43","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 17:33:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17ac14865127463ab399430c8255e67b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/205008fb332248b39f9f5b0acfb10c63 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"11ff642fe688a35afa3876f0f9dd7c4425f13d95","revisions":{"ca5a215881f1cf88153781db11701af5d29d03b2":{"kind":"REWORK","_number":1,"created":"2025-02-12 13:07:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/941388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/88/941388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/88/941388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/88/941388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/88/941388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/88/941388/1"}}},"commit":{"parents":[{"commit":"fd8e2522fc39e48da63d2137b1a8ccc7518636d2","subject":"Auto-fix yaml rules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/fd8e2522fc39e48da63d2137b1a8ccc7518636d2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-02-12 13:07:12.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-02-12 13:07:12.000000000","tz":60},"subject":"Auto-fix usage of modules via FQCN","message":"Auto-fix usage of modules via FQCN\n\nSince ansible-core 2.10 it is recommended to use modules via FQCN\nIn order to align with recommendation, we perform migration\nby applying suggestions made by `ansible-lint --fix\u003dfqcn`\n\nChange-Id: I335f0f1bcdb5e5564ce3f82f44eec7d8c6ab4e0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/ca5a215881f1cf88153781db11701af5d29d03b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/ca5a215881f1cf88153781db11701af5d29d03b2"}]},"branch":"refs/heads/master"},"be450968b72b8534bdb5ad14088e9d9da4a50bf0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-18 12:00:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/941388/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/88/941388/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/88/941388/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/88/941388/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/88/941388/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/88/941388/2"}}},"commit":{"parents":[{"commit":"aa1503d8cec501c2c3ed0357a9ed8ca0ad12f2cf","subject":"Auto-fix yaml rules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/aa1503d8cec501c2c3ed0357a9ed8ca0ad12f2cf"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-02-12 13:07:12.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-02-18 12:00:46.000000000","tz":0},"subject":"Auto-fix usage of modules via FQCN","message":"Auto-fix usage of modules via FQCN\n\nSince ansible-core 2.10 it is recommended to use modules via FQCN\nIn order to align with recommendation, we perform migration\nby applying suggestions made by `ansible-lint --fix\u003dfqcn`\n\nChange-Id: I335f0f1bcdb5e5564ce3f82f44eec7d8c6ab4e0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/be450968b72b8534bdb5ad14088e9d9da4a50bf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/be450968b72b8534bdb5ad14088e9d9da4a50bf0"}]},"branch":"refs/heads/master","description":"Rebase"},"11ff642fe688a35afa3876f0f9dd7c4425f13d95":{"kind":"REWORK","_number":3,"created":"2025-02-28 08:38:53.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/941388/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/88/941388/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/88/941388/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/88/941388/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/88/941388/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/88/941388/3"}}},"commit":{"parents":[{"commit":"aa1503d8cec501c2c3ed0357a9ed8ca0ad12f2cf","subject":"Auto-fix yaml rules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/aa1503d8cec501c2c3ed0357a9ed8ca0ad12f2cf"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-02-12 13:07:12.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-02-28 08:38:53.000000000","tz":0},"subject":"Auto-fix usage of modules via FQCN","message":"Auto-fix usage of modules via FQCN\n\nSince ansible-core 2.10 it is recommended to use modules via FQCN\nIn order to align with recommendation, we perform migration\nby applying suggestions made by `ansible-lint --fix\u003dfqcn`\n\nChange-Id: I335f0f1bcdb5e5564ce3f82f44eec7d8c6ab4e0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/11ff642fe688a35afa3876f0f9dd7c4425f13d95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/11ff642fe688a35afa3876f0f9dd7c4425f13d95"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
