)]}'
{"id":"openstack%2Fopenstack-ansible-os_nova~991981","triplet_id":"openstack%2Fopenstack-ansible-os_nova~master~I45e3fbdd6985e2d05de7ee464a6f6c900b4d6836","project":"openstack/openstack-ansible-os_nova","branch":"master","topic":"osa/policy_yaml","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-08-25 11:14:50.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-29 20:31:14.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-09-30 00:44:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I45e3fbdd6985e2d05de7ee464a6f6c900b4d6836","subject":"Update details about policy overrides","status":"MERGED","created":"2026-06-06 18:14:35.000000000","updated":"2026-09-30 00:45:15.000000000","submitted":"2026-09-30 00:44:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991981-osa/policy_yaml","meta_rev_id":"f7529994abbb8640a8ac9e17dddcb79aa0a26fc8","_number":991981,"virtual_id_number":991981,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-30 00:44:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-09-28 09:16:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2026-08-25 11:14:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2026-09-28 09:16:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-06 18:54:44.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-06-06 20:42:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-01 10:52:47.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-09-28 09:16:39.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"478319ab2edf6bd864ab34655e21b15e293de400","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-06 18:14:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be6f13634e8b3f1e303ea04e39fae102d8bc4ea0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-06 18:21:06.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"5301291966f2091764010c4e6cbb5743b8ba4b06","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-06 18:21:13.000000000","message":"Topic set to osa/policy_yaml","accounts_in_message":[],"_revision_number":2},{"id":"343b0e8e97c314f4a155e6a8a806d1c39c3be5d8","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-06 18:21:15.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"a7934e94a57f7c5ea1af47c95659782688a8ee8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 20:42:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99e4f622da8942a7b2dc49276aefaf2f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d49e2fdc5c4d483ea019d0cc66e9638b : FAILURE in 9m 19s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/79ed730c8a154a5ba2623267976db719 : SUCCESS in 1h 28m 32s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/66202c64a73d47699b2b16e8f9230e15 : SUCCESS in 1h 06m 43s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/e455702fb0234231a21915d8553e3a58 : SUCCESS in 1h 23m 44s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/df5a885b1762463f8d24a30420a4fc00 : SUCCESS in 48m 43s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/830a19161bc94d48ab64c79f2c99ba9e : SUCCESS in 1h 16m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec28d227f8554e8bb06b1a92afb76005 : SUCCESS in 1h 14m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e598faf1f44d4a559a2ea25a6004fea8 : SUCCESS in 1h 27m 25s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edf36dde0b0141f0aa0b2b004e603a1a : SUCCESS in 1h 40m 40s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3a8e4d98adb34c428743a5d731f722c1 : SUCCESS in 2h 17m 10s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c829ac28002e49f5bd1803e725cf9aad : SUCCESS in 1h 41m 07s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df11a10b4d7044c9a11dcd9fb3e4e41b : SUCCESS in 2h 08m 10s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e24af1ad1863458588e1cbe55cc766cf : SUCCESS in 2h 14m 16s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e72d8ff2ea14e948590d20cf130e1a1 : SUCCESS in 1h 39m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c2d5cd73abb492998880d57702ed228 : SUCCESS in 1h 22m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/adfe6ecfc3e346e89c354b6b21a023b1 : FAILURE in 20m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dc5a363a7804ea0935e5c2b22121a24 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c802bad8e5714cb9ae7a4447709cfc8c : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":2},{"id":"ebff4518ecaedd5b03e5210334ef7cbe6cac8df7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-06 22:47:12.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"07b839e00c6b76aa02138d6f9e84810db6b49f1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-07 01:20:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccdc3266c17d42ef8d4df7b62e98af67\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eefbc3cd300e40a4a5efe12ea2e3e34e : SUCCESS in 13m 43s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/7e6e3317a5e6403c9eebed68d0bcd2c9 : SUCCESS in 1h 27m 23s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3d0d58503d3d4b6aae32870d8a159bfe : SUCCESS in 1h 26m 28s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/5bd386286fa9431089ca752a8369db76 : SUCCESS in 1h 32m 24s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/48362b9ce7214d2a83dc9b1a524e983e : SUCCESS in 54m 43s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2ef9c47e2e1d456ea8ec36c179134ca6 : SUCCESS in 56m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0c1297c0be24bea961752afe9693377 : SUCCESS in 1h 31m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d07a67046de64e51ab1974b1989669ad : SUCCESS in 1h 31m 16s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/137d7689f11042a49e9a4045e5aae09e : SUCCESS in 1h 39m 07s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7fa1aed699074386b92d0d9f1252b9d9 : SUCCESS in 2h 16m 24s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e3fe608f0a384dc69680a79f19883997 : SUCCESS in 1h 38m 45s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5990699790cf40c28d731e32d3ebd4d9 : SUCCESS in 2h 12m 29s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35a575529e3c4260a5a6295ee99b6dfa : SUCCESS in 2h 20m 34s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e8bf4b5657f4059ad3ee080095ec224 : SUCCESS in 2h 24m 35s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30e833576270452997fdf1c5cfeab644 : SUCCESS in 1h 22m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/9cf92b75203e4bc58c858c25fcb36bf2 : FAILURE in 22m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b4ac271025a41729aa4a9b2060863fc : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3613c8f5dac4b13a07f77f5c08467c1 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":3},{"id":"253bb504da21ae67d08f889cdea5cb7ee8d4c1c5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-24 08:17:09.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"99eb062de2e627e95499fc6082e06815f684a0a9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-01 12:53:01.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"212586f2804518f5e81573b2eabc3c38243da87a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 05:11:50.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"a57bc6380093bdd73fc374517b6779892619c2ed","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 05:11:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"50751cfe9e88576b8fa5591b4e5eb1e3d876640b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-25 08:43:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f570bfcfe8846a58d665a5916671543\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/948bd87230db493fa3f8fbeb22a9e448 : SUCCESS in 15m 13s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/4c478945498e4f02bf8f676b366d9473 : TIMED_OUT in 3h 14m 05s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/801f3993e9e448e5a14467a9c4e8f735 : SUCCESS in 1h 35m 10s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/261129be62b44d63b188313d780cf38c : SUCCESS in 1h 37m 19s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e7c88a9e957f4a98b7c93f4884a3c6bc : SUCCESS in 1h 24m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43e8ecf9fe494e80b32f0f9d69d9b9cb : SUCCESS in 1h 37m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/68c28dd213a5431a8b5f919133004e9f : SUCCESS in 1h 36m 03s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19f6b9a580a4497582538490530f1e66 : SUCCESS in 1h 37m 07s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6744f7961dc743cb847c092ec84f23b4 : SUCCESS in 2h 07m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/718ae71d220c40188df45aa8fb95c6b6 : SUCCESS in 2h 06m 01s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1ee0d3a15bb452599856e9a740c4514 : SUCCESS in 2h 36m 17s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cd6a047fa0f46bd99fc3678daaea15d : SUCCESS in 1h 30m 46s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f416b0652ddc4036bf3541cfe7c8e19a : SUCCESS in 1h 23m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96c5bced973f44e982408ddc7a11709b : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b18de07a4bc4499861d76dd284562f1 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":4},{"id":"9dad42e00cbbd821ee8a03b933e4968181366af1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 08:55:25.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"dfc161675a05815f20291762ea0bb23c32826539","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-25 11:14:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/782f9c030dee4f83baace099afeb3bed\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/952431950f424c0b81dd3ccd788abc67 : SUCCESS in 10m 33s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/47a9f7f6091c4719942ac95e95c7a9e7 : SUCCESS in 2h 14m 40s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/4bd81604ba574924be474ff4600b45e8 : SUCCESS in 49m 34s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dc0c4a62a0024126946f42b2316d3084 : SUCCESS in 59m 44s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/50313db9d7fe460c97d37c038c351f4d : SUCCESS in 1h 19m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/167527d362ce48beb2d42baef28f6d0e : SUCCESS in 53m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/1b4c8ae51b154b898d17a8bf7ac7f8b3 : SUCCESS in 52m 37s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8730d3c561b4443b6578ff08ea6a438 : SUCCESS in 1h 17m 45s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0bed792647c64b47996435c7d5132617 : SUCCESS in 1h 20m 26s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fbd4d1291e8742c883fd8bae514799fd : SUCCESS in 1h 18m 00s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ddee70a87724299bf4c7d544a8cebfb : SUCCESS in 1h 27m 18s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b570c1e4caea4d1a974960413ba18fa6 : SUCCESS in 1h 02m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e0c34e86837444f5b887ddd04b932e06 : SUCCESS in 1h 21m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cabb197ba1e469db116fe94813495c2 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dec4fa5e48594f6183522a2e6f8021d1 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":5},{"id":"7a86479c8aed256d578ca08ac0dc9aebdabb566a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-25 11:14:50.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f6f9c3e09d6820541ea1b6d7f963e65980bdffeb","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-09-28 09:16:39.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"62e4d96d7264845df97f97891039045cff553ee9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 09:24:04.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"962806aa84307eb01145104799a668c575532da7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 11:28:27.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/23a8d567b63c42799cf4d5c86da559fa\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/658fefb373c145be9cf1efc4f994396e : SUCCESS in 14m 09s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/430f9bbe32f747609696cc0d6e8a96ad : SUCCESS in 52m 31s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/4a63329982a742f4b949274299cb33d4 : SUCCESS in 1h 30m 03s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/47bb5f3fa29a46e3a36b7d84eedcd460 : SUCCESS in 1h 31m 26s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/42e5f3b9ad80405f957e5e7001e51dd3 : SUCCESS in 55m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0bf618bc1fb45e4a870528be34be8da : SUCCESS in 1h 36m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/5c8d7958e80f4927bdff5ebaaed86f30 : SUCCESS in 1h 38m 27s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4815ae387a64448e8a38603b537b0b15 : FAILURE in 52m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b93f6ebe78c4a6fb2398a11c5c2278e : SUCCESS in 58m 12s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f2885300da1f4fa9841b3a8b86d43a6d : SUCCESS in 1h 24m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0657426e39a54cc88b6b65380038049a : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bc1d4d8735f4f3a8975b978a1adb073 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":5},{"id":"6ad398e3cea7cf8889110a65e7d2d48177500ad6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-29 20:31:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8679cced274f5d3cfeb5aaa8f5541b738245bac9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 23:05:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd2116a11a014498822accd0b79a45d3\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b7f199f2bca42d983578b4b5dc70bfa : SUCCESS in 11m 59s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/07342457698946a89cdb316c1c5377be : SUCCESS in 1h 11m 41s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/fb09b186e2cf45c29dbb5da56b4f8290 : SUCCESS in 1h 19m 48s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9379eddb7ae34f52bcafe51b119f6a24 : SUCCESS in 1h 18m 22s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/42fa6b6033bf4ce4a89f4a24a810d940 : SUCCESS in 51m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d404b2b8ae1f4af1a5be1cbfadac4ff7 : SUCCESS in 1h 51m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/dde96db4178d442a84937c3f6466a714 : SUCCESS in 1h 17m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6cb467e9b26451eac18832094161448 : SUCCESS in 1h 27m 54s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/71013a38f4904259af4670e2dd459c02 : SUCCESS in 1h 21m 09s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a75750350f22426d95a83ee26c6f7ebd : SUCCESS in 2h 17m 49s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f55c7e9c52824d0abb047dcf0a2b194c : SUCCESS in 2h 19m 16s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be484e1d56154f609450e2262e7eb279 : SUCCESS in 1h 18m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/ccc468040e8f4545845d7e1200bbc993 : SUCCESS in 51m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e9d47e356f047519e1c0da5991d1b38 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0f1ec06b7b14177bbd6bfa6104af512 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":5},{"id":"bf385d4d197f6984a24cd17667502a688b33aeeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 23:07:43.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f62a1ad3ecfdd52e0abada2ef7d511062429b92a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 00:44:15.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac1266b6674a42f0874edccea8144b2e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aeab7e4be0cc4326b9aa4eae5a2c9a67 : SUCCESS in 10m 14s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/3f44c6b4ec474c008c561b04a03b753b : SUCCESS in 45m 09s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/8ac094245e314b6f85bcb07405d43aa9 : SUCCESS in 1h 21m 30s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e7109b6f798244f2939d997da9121c29 : SUCCESS in 1h 34m 39s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/17e7013c21e24c589eda4d2dffdc0c52 : SUCCESS in 49m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6dfadbba0b74bc0abef090c508100a0 : SUCCESS in 1h 20m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/af48b7adbdb4473ab4f70317d71fd7d7 : SUCCESS in 1h 20m 17s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40c46508ca9e480f85c8b0d143f9a2ca : SUCCESS in 46m 05s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c674a12a9f7740b09d3947ef3908d345 : SUCCESS in 1h 20m 26s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/eba47f974edf4b15b815eb7164c3c6a1 : SUCCESS in 1h 21m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cffe4a9eb0ac44319489d7c6fe955888 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4909d48106374c1593630dddeefc83fa : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":5},{"id":"fc5b80f3967305448561c9501866024f827ec0dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 00:44:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"f7529994abbb8640a8ac9e17dddcb79aa0a26fc8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 00:45:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f983bbebd584ae28f2635912e7d71a5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53bde683e72d4bf8a87772afbc509fa5 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"fd9a34aeb21b841b53c98f255ed548d9b475c5ac","revisions":{"338625af9fe4d0362a6edbaf0bb3e40cbed84ce6":{"kind":"REWORK","_number":1,"created":"2026-06-06 18:14:35.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/81/991981/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/81/991981/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/81/991981/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/81/991981/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/81/991981/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/81/991981/1"}}},"commit":{"parents":[{"commit":"87fec8b45578cc232b2c4f152073118f102a97fa","subject":"Fix permissions on vencrypt console client private key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/87fec8b45578cc232b2c4f152073118f102a97fa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:14:35.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:14:35.000000000","tz":0},"subject":"Update details about policy overrides","message":"Update details about policy overrides\n\nChange-Id: I45e3fbdd6985e2d05de7ee464a6f6c900b4d6836\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/338625af9fe4d0362a6edbaf0bb3e40cbed84ce6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/338625af9fe4d0362a6edbaf0bb3e40cbed84ce6"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"80e0a8136493e96c1b71a98d0a6ce84299ce4650":{"kind":"REWORK","_number":2,"created":"2026-06-06 18:21:06.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/81/991981/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/81/991981/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/81/991981/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/81/991981/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/81/991981/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/81/991981/2"}}},"commit":{"parents":[{"commit":"87fec8b45578cc232b2c4f152073118f102a97fa","subject":"Fix permissions on vencrypt console client private key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/87fec8b45578cc232b2c4f152073118f102a97fa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:14:35.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:21:06.000000000","tz":0},"subject":"Update details about policy overrides","message":"Update details about policy overrides\n\nChange-Id: I45e3fbdd6985e2d05de7ee464a6f6c900b4d6836\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/80e0a8136493e96c1b71a98d0a6ce84299ce4650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/80e0a8136493e96c1b71a98d0a6ce84299ce4650"}]},"branch":"refs/heads/master"},"f48c670dde99d1fd0e53d91ff9a4877a3f6ba2f8":{"kind":"REWORK","_number":3,"created":"2026-06-06 22:47:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/81/991981/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/81/991981/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/81/991981/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/81/991981/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/81/991981/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/81/991981/3"}}},"commit":{"parents":[{"commit":"87fec8b45578cc232b2c4f152073118f102a97fa","subject":"Fix permissions on vencrypt console client private key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/87fec8b45578cc232b2c4f152073118f102a97fa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:14:35.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 22:47:12.000000000","tz":0},"subject":"Update details about policy overrides","message":"Update details about policy overrides\n\nChange-Id: I45e3fbdd6985e2d05de7ee464a6f6c900b4d6836\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/f48c670dde99d1fd0e53d91ff9a4877a3f6ba2f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/f48c670dde99d1fd0e53d91ff9a4877a3f6ba2f8"}]},"branch":"refs/heads/master"},"6aae7b45e304746325c4dac8280a1d1fba250fe3":{"kind":"REWORK","_number":4,"created":"2026-07-25 05:11:50.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/81/991981/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/81/991981/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/81/991981/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/81/991981/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/81/991981/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/81/991981/4"}}},"commit":{"parents":[{"commit":"87fec8b45578cc232b2c4f152073118f102a97fa","subject":"Fix permissions on vencrypt console client private key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/87fec8b45578cc232b2c4f152073118f102a97fa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:14:35.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 05:11:47.000000000","tz":0},"subject":"Update details about policy overrides","message":"Update details about policy overrides\n\nChange-Id: I45e3fbdd6985e2d05de7ee464a6f6c900b4d6836\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/6aae7b45e304746325c4dac8280a1d1fba250fe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/6aae7b45e304746325c4dac8280a1d1fba250fe3"}]},"branch":"refs/heads/master"},"fd9a34aeb21b841b53c98f255ed548d9b475c5ac":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-07-25 08:55:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/81/991981/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_nova","ref":"refs/changes/81/991981/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/81/991981/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/81/991981/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/81/991981/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_nova refs/changes/81/991981/5"}}},"commit":{"parents":[{"commit":"e06c7bceef3f7c365828116b7aadea6fc2d3e22b","subject":"Merge \"Ensure pyroute2 is being installed in venv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/e06c7bceef3f7c365828116b7aadea6fc2d3e22b"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:14:35.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 08:55:25.000000000","tz":0},"subject":"Update details about policy overrides","message":"Update details about policy overrides\n\nChange-Id: I45e3fbdd6985e2d05de7ee464a6f6c900b4d6836\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/fd9a34aeb21b841b53c98f255ed548d9b475c5ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_nova/commit/fd9a34aeb21b841b53c98f255ed548d9b475c5ac"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"87fec8b45578cc232b2c4f152073118f102a97fa","ours":"6aae7b45e304746325c4dac8280a1d1fba250fe3","theirs":"e06c7bceef3f7c365828116b7aadea6fc2d3e22b","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
