)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~587739","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I15b482fcd82e9b21028a417652f4d00a2b64c84b","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"move-mq-setup-into-roles","hashtags":[],"change_id":"I15b482fcd82e9b21028a417652f4d00a2b64c84b","subject":"Correct group names for services","status":"MERGED","created":"2018-08-01 09:31:41.000000000","updated":"2019-07-23 19:15:03.000000000","submitted":"2018-08-01 22:55:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"587739-1533164109301-4363ebbb","meta_rev_id":"974cffddda309704a1299b93f6b48ea9d1fb6682","_number":587739,"virtual_id_number":587739,"owner":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2018-08-01 22:55:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2018-08-01 19:59:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2018-08-01 19:59:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2018-08-01 19:59:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2018-08-01 19:59:19.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-08-01 13:50:08.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2018-08-01 13:50:26.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2018-08-01 15:38:50.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2018-08-01 22:55:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47b53964bee4b0c0dd70806ad5f5e9fca91fdc6b","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2018-08-01 09:31:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0d435e1d08959d3e31f0539fc7f84a55537bfd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-01 10:47:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/39/587739/1/check/build-openstack-sphinx-docs/fee2436/html/ : SUCCESS in 3m 17s\n- openstack-ansible-linters http://logs.openstack.org/39/587739/1/check/openstack-ansible-linters/7270977/ : SUCCESS in 6m 25s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/39/587739/1/check/openstack-ansible-functional-ubuntu-xenial/0a1cb53/ : FAILURE in 1h 10m 54s\n- openstack-ansible-octavia-ssl-nv http://logs.openstack.org/39/587739/1/check/openstack-ansible-octavia-ssl-nv/830ab84/ : FAILURE in 1h 05m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5a6a0d38a42319be7a399f2a9f5bbf82015bccb","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2018-08-01 11:57:13.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3551f387b13718b85a5a38dddb682b8728df4745","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-01 13:31:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/39/587739/2/check/build-openstack-sphinx-docs/41a2571/html/ : SUCCESS in 3m 40s\n- openstack-ansible-linters http://logs.openstack.org/39/587739/2/check/openstack-ansible-linters/4abe452/ : SUCCESS in 5m 57s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/39/587739/2/check/openstack-ansible-functional-ubuntu-xenial/475bcb9/ : SUCCESS in 1h 27m 26s\n- openstack-ansible-octavia-ssl-nv http://logs.openstack.org/39/587739/2/check/openstack-ansible-octavia-ssl-nv/61cd564/ : SUCCESS in 1h 19m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e9546e2b19b807ca91fa0bb1119e424878968698","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2018-08-01 13:50:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9ea1b74145435dd5ad42c80d100649c3f86e0552","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-08-01 13:50:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"08210ac0a9192f79dfc7815c6389e5e6070c5d30","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2018-08-01 15:38:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bbf9fbf5bb467e747a252ba42a2e1db15cf50c01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-01 15:39:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"68d036f5dd9b24d1ecd386704961b5b0aa0741b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-01 16:38:10.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"1209d284d8be11ecf72165461f5e18707e243f90","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2018-08-01 19:59:16.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c0da0849d5dc0e18c1d346738bb2169d2edbda3b","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2018-08-01 19:59:19.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"aba3279cae44c44acf00484c2943c5c5aaac1746","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-01 21:26:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/39/587739/3/check/build-openstack-sphinx-docs/39bc14e/html/ : SUCCESS in 2m 46s\n- openstack-ansible-linters http://logs.openstack.org/39/587739/3/check/openstack-ansible-linters/ff71310/ : SUCCESS in 7m 03s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/39/587739/3/check/openstack-ansible-functional-ubuntu-xenial/9e69f43/ : SUCCESS in 1h 15m 36s\n- openstack-ansible-octavia-ssl-nv http://logs.openstack.org/39/587739/3/check/openstack-ansible-octavia-ssl-nv/d5edcc1/ : SUCCESS in 1h 19m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"875dc02791eebfedf0b43d37a7e7b536237c8f83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-01 21:26:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2fde64174bec2cb5d26bc314fddc7d0f5ba18b96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-01 22:55:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"bdabf2ad532ce031cd4d6ba320aa550fa2fd89b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-01 22:55:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/39/587739/3/gate/build-openstack-sphinx-docs/75ee289/html/ : SUCCESS in 3m 15s\n- openstack-ansible-linters http://logs.openstack.org/39/587739/3/gate/openstack-ansible-linters/0e9bbe5/ : SUCCESS in 6m 09s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/39/587739/3/gate/openstack-ansible-functional-ubuntu-xenial/45b2f66/ : SUCCESS in 1h 20m 00s","accounts_in_message":[],"_revision_number":3},{"id":"55f3a08c96819f6eabea5ad511f0ba27ed11b5ab","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-07-23 19:15:03.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nCorrect group name is octavia-health-manager","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a00270d551f084aca77464193ad16a184adc7bf0","revisions":{"7c5cb923db49463be8fae7ad43c03eaa83443533":{"kind":"REWORK","_number":1,"created":"2018-08-01 09:31:41.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/39/587739/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/39/587739/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/39/587739/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/39/587739/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/39/587739/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/39/587739/1"}}},"commit":{"parents":[{"commit":"aedc45d472d2afa459533bcdcaf9bdf66fb92a35","subject":"Move MQ vhost/user creation into role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/aedc45d472d2afa459533bcdcaf9bdf66fb92a35"}]}],"author":{"name":"Jesse Pretorius","email":"jesse.pretorius@rackspace.co.uk","date":"2018-08-01 09:30:28.000000000","tz":60},"committer":{"name":"Jesse Pretorius","email":"jesse.pretorius@rackspace.co.uk","date":"2018-08-01 09:30:28.000000000","tz":60},"subject":"Correct group names for services","message":"Correct group names for services\n\nThe group names are incorrecly using underscores instead\nof hyphens. Even though using hyphens in group names is\na bad practise, it\u0027s already established that way in the\nintegrated repo, so we change these to match those.\n\nChange-Id: I15b482fcd82e9b21028a417652f4d00a2b64c84b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7c5cb923db49463be8fae7ad43c03eaa83443533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7c5cb923db49463be8fae7ad43c03eaa83443533"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"aedc45d472d2afa459533bcdcaf9bdf66fb92a35","is_merged_in_target_branch":false,"change_id":"If7e3712dd70eb083b5dffc98e0c981ec4f513533","change_number":583388,"patch_set_number":8,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ab2485d68cc6d58e9cf37b988513d6437f595e32":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2018-08-01 11:57:13.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/39/587739/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/39/587739/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/39/587739/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/39/587739/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/39/587739/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/39/587739/2"}}},"commit":{"parents":[{"commit":"06a044d487a83bb6ef7ce82f5266500b1e741a4d","subject":"Move MQ vhost/user creation into role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/06a044d487a83bb6ef7ce82f5266500b1e741a4d"}]}],"author":{"name":"Jesse Pretorius","email":"jesse.pretorius@rackspace.co.uk","date":"2018-08-01 09:30:28.000000000","tz":60},"committer":{"name":"Jesse Pretorius (odyssey4me)","email":"jesse.pretorius@rackspace.co.uk","date":"2018-08-01 11:57:13.000000000","tz":0},"subject":"Correct group names for services","message":"Correct group names for services\n\nThe group names are incorrecly using underscores instead\nof hyphens. Even though using hyphens in group names is\na bad practise, it\u0027s already established that way in the\nintegrated repo, so we change these to match those.\n\nChange-Id: I15b482fcd82e9b21028a417652f4d00a2b64c84b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ab2485d68cc6d58e9cf37b988513d6437f595e32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ab2485d68cc6d58e9cf37b988513d6437f595e32"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"06a044d487a83bb6ef7ce82f5266500b1e741a4d","is_merged_in_target_branch":false,"change_id":"If7e3712dd70eb083b5dffc98e0c981ec4f513533","change_number":583388,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a00270d551f084aca77464193ad16a184adc7bf0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2018-08-01 19:59:16.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/39/587739/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/39/587739/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/39/587739/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/39/587739/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/39/587739/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/39/587739/3"}}},"commit":{"parents":[{"commit":"b6d495c9c9f7cf6181414c218d5e1cd24615a179","subject":"Move MQ vhost/user creation into role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b6d495c9c9f7cf6181414c218d5e1cd24615a179"}]}],"author":{"name":"Jesse Pretorius","email":"jesse.pretorius@rackspace.co.uk","date":"2018-08-01 09:30:28.000000000","tz":60},"committer":{"name":"Jesse Pretorius (odyssey4me)","email":"jesse.pretorius@rackspace.co.uk","date":"2018-08-01 19:59:16.000000000","tz":0},"subject":"Correct group names for services","message":"Correct group names for services\n\nThe group names are incorrecly using underscores instead\nof hyphens. Even though using hyphens in group names is\na bad practise, it\u0027s already established that way in the\nintegrated repo, so we change these to match those.\n\nChange-Id: I15b482fcd82e9b21028a417652f4d00a2b64c84b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a00270d551f084aca77464193ad16a184adc7bf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a00270d551f084aca77464193ad16a184adc7bf0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b6d495c9c9f7cf6181414c218d5e1cd24615a179","is_merged_in_target_branch":true,"change_id":"If7e3712dd70eb083b5dffc98e0c981ec4f513533","change_number":583388,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
