)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~635402","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I3ba16015e427afb92c8e99db3d4b3d311d7074df","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"change635402","hashtags":[],"change_id":"I3ba16015e427afb92c8e99db3d4b3d311d7074df","subject":"Update octavia configuration documentation","status":"MERGED","created":"2019-02-07 03:41:24.000000000","updated":"2020-02-25 17:40:42.000000000","submitted":"2020-02-25 17:38:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"635402-1582652311123-20d59637","meta_rev_id":"bda5398fa33276c87b3620c5dced7f3cfe509692","_number":635402,"virtual_id_number":635402,"owner":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-25 17:38:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-22 00:52:17.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2020-02-25 17:19:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2020-02-25 17:19:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-07 04:51:23.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-02-07 14:37:03.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-02-07 21:04:50.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2020-02-22 00:52:17.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-02-25 17:19:45.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-02-25 17:38:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f0df28aeafcf822b476c6302a8c26e97ef700bb","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-02-07 03:41:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9922a3be85522c888cf87458b047ab342e63289d","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-02-07 03:42:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0e5723708cb44402251d9d9cc713ab930fca0649","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-07 03:50:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/02/635402/2/check/openstack-tox-docs/88e558b/html/ : SUCCESS in 2m 15s\n- openstack-ansible-linters http://logs.openstack.org/02/635402/2/check/openstack-ansible-linters/9c9348b/ : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":2},{"id":"df25a7f8bcdce538f7c16d9475f64778fabe7e0c","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-02-07 04:51:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"34d00d0c26b3a8ad0cec2da059a73c73ca19b6ff","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-02-07 14:37:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3fb00f9bf26d49c316e656dc99664360ee271fd2","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-02-07 16:48:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6045ff14f6a5368cd78e231ed31e7bd535effeb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-07 17:06:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/02/635402/3/check/openstack-tox-docs/8e65bb7/html/ : SUCCESS in 3m 06s\n- openstack-ansible-linters http://logs.openstack.org/02/635402/3/check/openstack-ansible-linters/0c3b4c8/ : SUCCESS in 9m 14s","accounts_in_message":[],"_revision_number":3},{"id":"783d5bdb7c6a657591e64bb348b2bb4839d330e8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-02-07 19:40:00.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d55aca9688ca075e3377f077403aa594c76f09ed","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-02-07 20:26:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"21fcda7935f9730f2401378c7013e39b2164a536","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-02-14 23:01:53.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"85cebfc1a0664ccec999bffc7e3743aae087e0d6","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2020-02-21 21:30:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"df9e2401d5d6efca05c48a4de96215d5e54e88d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 21:42:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c49fb3c7aad34442bf3766cba184c9c5 : SUCCESS in 9m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8652f858a094d2eb1c6db0e98fba783 : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":4},{"id":"c6dc4e7c22b57e11466cad7cfeb81dfeab21d4c1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-22 00:52:17.000000000","message":"Patch Set 4: Code-Review+2\n\nBetter than it was, thanks Bjoern","accounts_in_message":[],"_revision_number":4},{"id":"0e5b2005032cfaf800c4f82f2cd09d54f980ec44","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-25 17:19:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5ad7041bc10b9beef4cac4c232355200e121521c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 17:19:58.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c0d988d3208e99148275718f7b4c37ac8c1dbcc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 17:38:30.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/eec14bb06d71472cba8ef9120269f57b : SUCCESS in 9m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e247f01e61124e1eb39d91196220581f : SUCCESS in 7m 36s","accounts_in_message":[],"_revision_number":4},{"id":"61cc4044f32e3cfaaa77d36f87ebef336e366cba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 17:38:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"90578428f8c00b6893182f3bb9628c1699652657","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 17:40:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d516ff5ef614e7fbf26cd4f790ba08b : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b20d64bdc356f9a6f54dd45b4e90dd4c1adbe7c4","revisions":{"d425c17027b2cc6fb6ba83427e0699d7ef780594":{"kind":"REWORK","_number":1,"created":"2019-02-07 03:41:24.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/02/635402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/02/635402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/02/635402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/02/635402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/02/635402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/02/635402/1"}}},"commit":{"parents":[{"commit":"eee659d342644de4fc87d15522d2e27f6d3a589e","subject":"Merge \"Add bugs url link to README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/eee659d342644de4fc87d15522d2e27f6d3a589e"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-02-07 03:39:00.000000000","tz":-360},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-02-07 03:41:23.000000000","tz":-360},"subject":"Update octavia configuration documentation","message":"Update octavia configuration documentation\n\nThe Configuring the Octavia Load Balancing service documentation\nis updated for Queens and upwards along with isolating pip package\ninstallations into a virtual environment\n\nChange-Id: I3ba16015e427afb92c8e99db3d4b3d311d7074df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/d425c17027b2cc6fb6ba83427e0699d7ef780594"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/d425c17027b2cc6fb6ba83427e0699d7ef780594"}]},"branch":"refs/heads/master"},"ac469a8da7b14001af6f65c3740aabc76a4561b4":{"kind":"REWORK","_number":2,"created":"2019-02-07 03:42:29.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/02/635402/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/02/635402/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/02/635402/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/02/635402/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/02/635402/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/02/635402/2"}}},"commit":{"parents":[{"commit":"eee659d342644de4fc87d15522d2e27f6d3a589e","subject":"Merge \"Add bugs url link to README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/eee659d342644de4fc87d15522d2e27f6d3a589e"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-02-07 03:39:00.000000000","tz":-360},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-02-07 03:42:23.000000000","tz":-360},"subject":"Update octavia configuration documentation","message":"Update octavia configuration documentation\n\nThe Configuring the Octavia Load Balancing service documentation\nis updated for Queens and upwards along with isolating pip package\ninstallations into a virtual environment\n\nChange-Id: I3ba16015e427afb92c8e99db3d4b3d311d7074df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ac469a8da7b14001af6f65c3740aabc76a4561b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ac469a8da7b14001af6f65c3740aabc76a4561b4"}]},"branch":"refs/heads/master"},"7ddd240a1ada31952641262787f2c72da61644b5":{"kind":"REWORK","_number":3,"created":"2019-02-07 16:48:27.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/02/635402/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/02/635402/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/02/635402/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/02/635402/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/02/635402/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/02/635402/3"}}},"commit":{"parents":[{"commit":"eee659d342644de4fc87d15522d2e27f6d3a589e","subject":"Merge \"Add bugs url link to README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/eee659d342644de4fc87d15522d2e27f6d3a589e"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-02-07 03:39:00.000000000","tz":-360},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-02-07 16:48:17.000000000","tz":-360},"subject":"Update octavia configuration documentation","message":"Update octavia configuration documentation\n\nThe Configuring the Octavia Load Balancing service documentation\nis updated for Queens and upwards along with isolating pip package\ninstallations into a virtual environment\n\nChange-Id: I3ba16015e427afb92c8e99db3d4b3d311d7074df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7ddd240a1ada31952641262787f2c72da61644b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7ddd240a1ada31952641262787f2c72da61644b5"}]},"branch":"refs/heads/master"},"b20d64bdc356f9a6f54dd45b4e90dd4c1adbe7c4":{"kind":"REWORK","_number":4,"created":"2020-02-21 21:30:44.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/02/635402/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/02/635402/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/02/635402/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/02/635402/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/02/635402/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/02/635402/4"}}},"commit":{"parents":[{"commit":"128f6ea134045592a0978c30afe08dfe041162b9","subject":"Fix typo.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/128f6ea134045592a0978c30afe08dfe041162b9"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2020-02-21 21:30:16.000000000","tz":-360},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2020-02-21 21:30:16.000000000","tz":-360},"subject":"Update octavia configuration documentation","message":"Update octavia configuration documentation\n\nThe Configuring the Octavia Load Balancing service documentation\nis updated for Queens and upwards along with isolating pip package\ninstallations into a virtual environment\n\nChange-Id: I3ba16015e427afb92c8e99db3d4b3d311d7074df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b20d64bdc356f9a6f54dd45b4e90dd4c1adbe7c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b20d64bdc356f9a6f54dd45b4e90dd4c1adbe7c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
