)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~652049","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I7eb1cbe2c80ee889d2ae08dcfed6a19cc1bd3415","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"bp/openstack-distribution-packages","hashtags":[],"change_id":"I7eb1cbe2c80ee889d2ae08dcfed6a19cc1bd3415","subject":"Add support for using distribution packages for OpenStack services","status":"MERGED","created":"2019-04-12 11:59:00.000000000","updated":"2019-06-28 14:18:48.000000000","submitted":"2019-06-28 14:18:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652049-1561731528943-0a7135d0","meta_rev_id":"144d7eb164c49fb4ce3e2ad2deeca339cd324ec3","_number":652049,"virtual_id_number":652049,"owner":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"value":0,"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-28 14:18:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"value":0,"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-28 10:42:51.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-06-28 12:16:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"value":0,"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-06-28 12:16:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"value":0,"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-12 12:20:27.000000000","updated_by":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"reviewer":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"state":"REVIEWER"},{"updated":"2019-05-24 15:35:06.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-28 10:42:51.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-06-28 12:16:27.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-28 14:18:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"19455fc1c4ce865c080a6ce5117ede421d62fb28","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-04-12 11:59:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6316f415c583478b2f7ace875c4e967a3b3207dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 11:59:12.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"e96c2c1e169495678cfd246269dad78547d7135b","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-04-12 12:03:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"641690cd7e958cc5603042390351b21b0ea97d1c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 13:18:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/2/check/openstack-ansible-linters/9678d15/ : FAILURE in 2m 38s\n- openstack-tox-docs http://logs.openstack.org/49/652049/2/check/openstack-tox-docs/d0ed0e3/ : POST_FAILURE in 2m 08s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/2/check/build-openstack-releasenotes/5713583/html/ : SUCCESS in 2m 55s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/49/652049/2/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/bd948e5/ : FAILURE in 1h 05m 47s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/49/652049/2/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/5295842/ : FAILURE in 48m 09s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/2/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/85cfd84/ : FAILURE in 42m 24s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/49/652049/2/check/openstack-ansible-functional-distro_install-centos-7/802c41c/ : FAILURE in 7m 28s\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/49/652049/2/check/openstack-ansible-functional-distro_install-ubuntu-bionic/5ad95fa/ : FAILURE in 3m 04s","accounts_in_message":[],"_revision_number":2},{"id":"1e2ed9f34d4c835cae9016f93c5b01312ba7ed3c","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-04-12 13:31:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"683784e5be5afa652ec450835ee404c8e5b8647f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-12 13:58:30.000000000","message":"Patch Set 3:\n\nI think it\u0027s not \"instead\", it\u0027s \"in addition\".","accounts_in_message":[],"_revision_number":3},{"id":"1f6c6648edd0fb7fbfb240ee30f87b91144b21ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 14:46:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/3/check/openstack-ansible-linters/07a57e9/ : SUCCESS in 6m 30s\n- openstack-tox-docs http://logs.openstack.org/49/652049/3/check/openstack-tox-docs/e00bb2e/html/ : SUCCESS in 2m 54s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/3/check/build-openstack-releasenotes/ee0bb89/html/ : SUCCESS in 2m 36s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/49/652049/3/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/c097ac8/ : FAILURE in 1h 10m 14s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/49/652049/3/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/d8b73ed/ : FAILURE in 49m 36s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/3/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/283a0f9/ : FAILURE in 49m 41s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/49/652049/3/check/openstack-ansible-functional-distro_install-centos-7/3c306e5/ : FAILURE in 17m 25s\n- openstack-ansible-functional-distro_install-ubuntu-bionic finger://ze11.openstack.org/fe58e6e1d0284f4ba9b41062c5e4fa7e : RETRY_LIMIT in 13m 44s","accounts_in_message":[],"_revision_number":3},{"id":"fd7763c49691244920d86efa77da7fac2be1177c","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-04-12 16:05:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7506431f8b704e31faec954e32793451a2f9931b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 17:35:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/4/check/openstack-ansible-linters/228955d/ : SUCCESS in 7m 07s\n- openstack-tox-docs http://logs.openstack.org/49/652049/4/check/openstack-tox-docs/05a42ac/html/ : SUCCESS in 3m 46s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/4/check/build-openstack-releasenotes/5ce8068/html/ : SUCCESS in 3m 36s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/49/652049/4/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/dbf0556/ : SUCCESS in 1h 23m 54s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/49/652049/4/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/4c5154d/ : SUCCESS in 1h 04m 03s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/4/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/35bd9db/ : SUCCESS in 1h 06m 15s\n- openstack-ansible-functional-distro_install-centos-7 finger://ze05.openstack.org/2758fde9e16343c38b9aab0103fbb9f6 : RETRY_LIMIT in 16m 56s\n- openstack-ansible-functional-distro_install-ubuntu-bionic finger://ze12.openstack.org/e624eec039e0477e9eae0fe8a98d820c : RETRY_LIMIT in 12m 58s","accounts_in_message":[],"_revision_number":4},{"id":"ebe776111230645fbd89659baff7b72c1a6cc2e9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-12 18:33:40.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"307515f6b5c14487721ea7b39235db037fb3afb9","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-04-12 19:07:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8b02301bb270830dd315ececcb3a8ed1459dccfb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 20:41:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/5/check/openstack-ansible-linters/247127f/ : SUCCESS in 6m 16s\n- openstack-tox-docs http://logs.openstack.org/49/652049/5/check/openstack-tox-docs/ecc7d05/html/ : SUCCESS in 2m 44s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/5/check/build-openstack-releasenotes/7312193/html/ : SUCCESS in 3m 29s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/49/652049/5/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/78272c0/ : SUCCESS in 1h 24m 22s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/49/652049/5/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/a8a6378/ : SUCCESS in 1h 09m 30s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/5/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/648a70c/ : SUCCESS in 1h 03m 21s\n- openstack-ansible-deploy-aio_distro_metal_octavia-centos-7 http://logs.openstack.org/49/652049/5/check/openstack-ansible-deploy-aio_distro_metal_octavia-centos-7/b826304/ : FAILURE in 41m 44s\n- openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable http://logs.openstack.org/49/652049/5/check/openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable/d2e65ac/ : FAILURE in 16m 33s\n- openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/5/check/openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic/6872c37/ : FAILURE in 35m 31s","accounts_in_message":[],"_revision_number":5},{"id":"33cc9ce7ae1c57622d12af3f28cf0e9ec9b4a309","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-04-12 21:02:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"dcb124ba7a284c458ffb22a18aa075112de5356f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 22:30:09.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/6/check/openstack-ansible-linters/8dff084/ : SUCCESS in 6m 02s\n- openstack-tox-docs http://logs.openstack.org/49/652049/6/check/openstack-tox-docs/f71f8dd/html/ : SUCCESS in 2m 54s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/6/check/build-openstack-releasenotes/6fd7a7d/html/ : SUCCESS in 2m 33s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/49/652049/6/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/eba5490/ : SUCCESS in 1h 22m 10s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/49/652049/6/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/4489f9f/ : SUCCESS in 1h 05m 27s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/6/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/d4fc3b1/ : SUCCESS in 1h 02m 21s\n- openstack-ansible-deploy-aio_distro_metal_octavia-centos-7 http://logs.openstack.org/49/652049/6/check/openstack-ansible-deploy-aio_distro_metal_octavia-centos-7/780d8bc/ : FAILURE in 46m 16s\n- openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable http://logs.openstack.org/49/652049/6/check/openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable/2fc901d/ : FAILURE in 19m 09s\n- openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/6/check/openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic/02e3420/ : FAILURE in 34m 45s","accounts_in_message":[],"_revision_number":6},{"id":"61d2a37599f4f211f370746175792565e31d9856","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-04-13 01:19:55.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1738582b1f836cc870e7c81cb377acea503cf4af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-13 02:55:07.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/7/check/openstack-ansible-linters/b47b358/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/49/652049/7/check/openstack-tox-docs/b72f3f2/html/ : SUCCESS in 2m 30s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/7/check/build-openstack-releasenotes/bb9a3de/html/ : SUCCESS in 2m 30s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/49/652049/7/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/b87ffed/ : SUCCESS in 1h 22m 45s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/49/652049/7/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/2d3bc73/ : SUCCESS in 52m 14s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/7/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/18a1191/ : SUCCESS in 1h 04m 01s\n- openstack-ansible-deploy-aio_distro_metal_octavia-centos-7 http://logs.openstack.org/49/652049/7/check/openstack-ansible-deploy-aio_distro_metal_octavia-centos-7/113ec7a/ : FAILURE in 42m 43s\n- openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable http://logs.openstack.org/49/652049/7/check/openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable/faadf16/ : FAILURE in 14m 44s\n- openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/7/check/openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic/8f091b0/ : SUCCESS in 48m 18s","accounts_in_message":[],"_revision_number":7},{"id":"c2fb0ec370e4b60f7534ec41695147e826b2790b","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-04-18 15:44:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3fcd2b08efa318c41c05adbeac34afdcad62ac55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 17:18:31.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/8/check/openstack-ansible-linters/93dd011/ : SUCCESS in 8m 13s\n- openstack-tox-docs http://logs.openstack.org/49/652049/8/check/openstack-tox-docs/ef1ac8c/html/ : SUCCESS in 2m 32s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/8/check/build-openstack-releasenotes/71a94e7/html/ : SUCCESS in 2m 20s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/49/652049/8/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/5779f53/ : SUCCESS in 1h 19m 53s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/49/652049/8/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/55691b6/ : SUCCESS in 1h 01m 24s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/8/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/b3f3116/ : SUCCESS in 1h 01m 56s\n- openstack-ansible-deploy-aio_distro_metal_octavia-centos-7 http://logs.openstack.org/49/652049/8/check/openstack-ansible-deploy-aio_distro_metal_octavia-centos-7/c923fd3/ : FAILURE in 41m 43s\n- openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable http://logs.openstack.org/49/652049/8/check/openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable/0106236/ : FAILURE in 18m 06s\n- openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/8/check/openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic/027c94d/ : FAILURE in 25m 52s","accounts_in_message":[],"_revision_number":8},{"id":"2b06bd799f0a4f9cea1912a0e4b977913cfecdef","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-04-20 12:07:06.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"3ef43c54b7dde606ff5a3dbda8c0934503749440","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 12:28:38.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/9/check/openstack-ansible-linters/8340071/ : FAILURE in 4m 25s\n- openstack-tox-docs http://logs.openstack.org/49/652049/9/check/openstack-tox-docs/5980144/html/ : SUCCESS in 2m 28s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/9/check/build-openstack-releasenotes/060a79d/html/ : SUCCESS in 2m 11s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/49/652049/9/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/0869b38/ : FAILURE in 14m 06s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/49/652049/9/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/7a23747/ : FAILURE in 9m 24s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/9/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/9ab3488/ : FAILURE in 10m 20s\n- openstack-ansible-deploy-aio_distro_metal_octavia-centos-7 http://logs.openstack.org/49/652049/9/check/openstack-ansible-deploy-aio_distro_metal_octavia-centos-7/dd4f9a9/ : FAILURE in 9m 06s\n- openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable http://logs.openstack.org/49/652049/9/check/openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable/77ad86b/ : FAILURE in 15m 24s\n- openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/9/check/openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic/9484062/ : FAILURE in 20m 41s","accounts_in_message":[],"_revision_number":9},{"id":"d6881cade097e293846eb3d815aa285ef0117312","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-22 16:58:44.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"3aad914b18fc36a3759b9481398a6534f4343782","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 19:11:35.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/9/check/openstack-ansible-linters/6098547/ : SUCCESS in 6m 15s\n- openstack-tox-docs http://logs.openstack.org/49/652049/9/check/openstack-tox-docs/ee9b1e9/html/ : SUCCESS in 2m 59s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/9/check/build-openstack-releasenotes/b440a71/html/ : SUCCESS in 2m 54s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/49/652049/9/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/acaf370/ : SUCCESS in 1h 36m 31s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/49/652049/9/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/cf0aa39/ : SUCCESS in 1h 03m 23s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/9/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/a8c6eaf/ : SUCCESS in 1h 00m 21s\n- openstack-ansible-deploy-aio_distro_metal_octavia-centos-7 http://logs.openstack.org/49/652049/9/check/openstack-ansible-deploy-aio_distro_metal_octavia-centos-7/935a862/ : FAILURE in 9m 59s\n- openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable http://logs.openstack.org/49/652049/9/check/openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable/365deea/ : FAILURE in 18m 30s\n- openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/9/check/openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic/214155b/ : FAILURE in 20m 47s","accounts_in_message":[],"_revision_number":9},{"id":"714df53d1f0a046cb05c86b6fda80387fb540de9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-22 19:23:05.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"0dfcc9eb7a3f725b79b4b05131aa255e0f41b3c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 20:44:12.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/9/check/openstack-ansible-linters/7a96609/ : SUCCESS in 4m 38s\n- openstack-tox-docs http://logs.openstack.org/49/652049/9/check/openstack-tox-docs/eff6f99/html/ : SUCCESS in 2m 18s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/9/check/build-openstack-releasenotes/37bca44/html/ : SUCCESS in 2m 13s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/49/652049/9/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/468df4f/ : SUCCESS in 1h 16m 56s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/49/652049/9/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/ecab118/ : SUCCESS in 1h 01m 42s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/9/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/511d647/ : SUCCESS in 1h 03m 36s\n- openstack-ansible-deploy-aio_distro_metal_octavia-centos-7 http://logs.openstack.org/49/652049/9/check/openstack-ansible-deploy-aio_distro_metal_octavia-centos-7/e9a6e4d/ : FAILURE in 9m 09s\n- openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable http://logs.openstack.org/49/652049/9/check/openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable/5edf23d/ : FAILURE in 17m 59s\n- openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/9/check/openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic/87a0b6a/ : FAILURE in 21m 07s","accounts_in_message":[],"_revision_number":9},{"id":"5b15127f1ea3c4fb7388cbf87ab475fc47763807","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-04-23 13:46:44.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"fbb2c6fd380873edfb6e95c7bab1286a250584c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 17:16:08.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/10/check/openstack-ansible-linters/a3fe5c0/ : SUCCESS in 6m 06s\n- openstack-tox-docs http://logs.openstack.org/49/652049/10/check/openstack-tox-docs/080147e/html/ : SUCCESS in 3m 15s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/10/check/build-openstack-releasenotes/1db36c0/html/ : SUCCESS in 3m 08s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 finger://ze06.openstack.org/058a2e104abe465b881aa18f9e926b2b : RETRY_LIMIT in 43m 48s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable finger://ze01.openstack.org/64858fcaef2a4d2fbf3716c99abde37a : RETRY_LIMIT in 43m 30s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic finger://ze04.openstack.org/966d3fbe3f214821a707af1cfb945915 : RETRY_LIMIT in 18m 27s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/49/652049/10/check/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/cce15bf/ : FAILURE in 9m 04s\n- openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable http://logs.openstack.org/49/652049/10/check/openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable/304997d/ : FAILURE in 20m 25s\n- openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/10/check/openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic/931f9a3/ : FAILURE in 21m 22s","accounts_in_message":[],"_revision_number":10},{"id":"6738cf1eeacd0d5ccda81c28250fdab50251d6a3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-25 16:53:01.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"4bb1bb65264fdc1b0af4a9c1f93d8773036979d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 18:23:03.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/10/check/openstack-ansible-linters/18a9c1d/ : SUCCESS in 8m 57s\n- openstack-tox-docs http://logs.openstack.org/49/652049/10/check/openstack-tox-docs/704302f/html/ : SUCCESS in 3m 47s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/10/check/build-openstack-releasenotes/6347204/html/ : SUCCESS in 2m 34s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/49/652049/10/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/59d2528/ : SUCCESS in 1h 23m 32s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/49/652049/10/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/2a10aff/ : SUCCESS in 1h 05m 18s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/10/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/690d05b/ : SUCCESS in 59m 39s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/49/652049/10/check/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/2d25e85/ : FAILURE in 9m 20s\n- openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable http://logs.openstack.org/49/652049/10/check/openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable/fc08969/ : FAILURE in 17m 42s\n- openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/10/check/openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic/4562284/ : FAILURE in 22m 06s","accounts_in_message":[],"_revision_number":10},{"id":"2814f7b1d79d9521f29204d3c7fc20fda53d44ed","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-05-06 17:42:54.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"235e1b893b01103f3dbdbd621c1995c2c92ad2c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 19:09:45.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/11/check/openstack-ansible-linters/e6b51d6/ : SUCCESS in 10m 56s\n- openstack-tox-docs http://logs.openstack.org/49/652049/11/check/openstack-tox-docs/7ad58c2/html/ : SUCCESS in 2m 49s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/11/check/build-openstack-releasenotes/1d2d07d/html/ : SUCCESS in 2m 03s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/49/652049/11/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/8856a69/ : SUCCESS in 1h 20m 47s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/49/652049/11/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/1a994a5/ : SUCCESS in 1h 02m 04s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/11/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/699fc63/ : SUCCESS in 1h 00m 18s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/49/652049/11/check/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/8a628cb/ : FAILURE in 9m 34s\n- openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable http://logs.openstack.org/49/652049/11/check/openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable/11d461e/ : FAILURE in 18m 46s\n- openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/11/check/openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic/06e3231/ : FAILURE in 19m 27s","accounts_in_message":[],"_revision_number":11},{"id":"ffd3fd07e1432828e84e8d4fac84fe4c570ae8a9","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-05-07 12:37:44.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"cc3a4ebaf974426f6bd928c12535903fb0cdc491","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-05-07 12:59:59.000000000","message":"Uploaded patch set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"85a38865b2a2abc02bb7ff20a0200687a2b2c908","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 14:13:39.000000000","message":"Patch Set 12: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"7b81981fb7d8e9c1317ebce2e402a65e9083c567","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-05-07 14:16:05.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"02c7f059886939a5c9fdc20e15e7ef9814ba7f0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 15:41:05.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/13/check/openstack-ansible-linters/a60c6b3/ : SUCCESS in 6m 46s\n- openstack-tox-docs http://logs.openstack.org/49/652049/13/check/openstack-tox-docs/999fe5f/html/ : SUCCESS in 2m 40s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/13/check/build-openstack-releasenotes/d4e2de0/html/ : SUCCESS in 2m 41s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/49/652049/13/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/d803a3e/ : SUCCESS in 1h 18m 57s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/49/652049/13/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/8ee2703/ : SUCCESS in 1h 07m 27s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/13/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/d1bd03d/ : SUCCESS in 1h 00m 43s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/49/652049/13/check/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/da6824b/ : FAILURE in 39m 21s\n- openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable http://logs.openstack.org/49/652049/13/check/openstack-ansible-deploy-aio_distro_metal_octavia-debian-stable/ea646ac/ : FAILURE in 18m 01s\n- openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/13/check/openstack-ansible-deploy-aio_distro_metal_octavia-ubuntu-bionic/a54b5ee/ : FAILURE in 22m 12s","accounts_in_message":[],"_revision_number":13},{"id":"0f031e2d12191d9f4af0b15dfadcbe1461021186","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-05-08 17:15:40.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"95fdbd593c346f2e011ad3ae52e64830ef3de723","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 18:41:27.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/14/check/openstack-ansible-linters/248b1ee/ : SUCCESS in 5m 56s\n- openstack-tox-docs http://logs.openstack.org/49/652049/14/check/openstack-tox-docs/141371a/html/ : SUCCESS in 2m 28s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/14/check/build-openstack-releasenotes/7ff8936/html/ : SUCCESS in 3m 15s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/49/652049/14/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/06961c6/ : SUCCESS in 1h 20m 37s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/49/652049/14/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/6d8dddb/ : SUCCESS in 1h 10m 34s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/14/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/e7e54d6/ : FAILURE in 1h 01m 55s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/49/652049/14/check/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/623d7a6/ : FAILURE in 36m 38s","accounts_in_message":[],"_revision_number":14},{"id":"4d95f9074bc3d890379e812c7c4ea56b91f44c96","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-05-09 14:50:06.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"038bec00e7ba5458aeb9cf47432148c8d6d3f1db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 16:21:05.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/14/check/openstack-ansible-linters/b617aa3/ : SUCCESS in 5m 48s\n- openstack-tox-docs http://logs.openstack.org/49/652049/14/check/openstack-tox-docs/9dea3e1/html/ : SUCCESS in 2m 14s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/14/check/build-openstack-releasenotes/8b3e2d9/html/ : SUCCESS in 3m 09s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/49/652049/14/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/c33d0ba/ : SUCCESS in 1h 25m 07s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/49/652049/14/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/e83c608/ : SUCCESS in 1h 09m 15s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/14/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/039b075/ : SUCCESS in 1h 01m 50s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/49/652049/14/check/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/9de9047/ : FAILURE in 43m 30s","accounts_in_message":[],"_revision_number":14},{"id":"843f7a682ca8e922987879c810f69606349e42f4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-24 15:35:06.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"fd170edeb286ad605b861a9d9f65596ad3532e29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 17:27:57.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/14/check/openstack-ansible-linters/8d9a6fa/ : SUCCESS in 4m 33s\n- openstack-tox-docs http://logs.openstack.org/49/652049/14/check/openstack-tox-docs/7a423e1/html/ : SUCCESS in 2m 46s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/14/check/build-openstack-releasenotes/2fa6195/html/ : SUCCESS in 2m 32s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/49/652049/14/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/68c676c/ : SUCCESS in 1h 43m 52s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/49/652049/14/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/b7f1c1d/ : FAILURE in 1h 18m 32s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/14/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/1f809a7/ : FAILURE in 1h 20m 14s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/49/652049/14/check/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/b755b9d/ : FAILURE in 44m 44s","accounts_in_message":[],"_revision_number":14},{"id":"10cc3aa32e439f3f55ce4eea16cc20ed88f925fb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 06:12:14.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"4c3977145bd754eca0b376b47ec184601ad83b59","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 08:55:06.000000000","message":"Patch Set 15: Commit message was updated.","accounts_in_message":[],"_revision_number":15},{"id":"3e6388589e07253ca674b2d5ed942f1e3970cee4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 12:11:25.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/15/check/openstack-ansible-linters/1b81cb9/ : SUCCESS in 6m 21s\n- openstack-tox-docs http://logs.openstack.org/49/652049/15/check/openstack-tox-docs/ada7578/html/ : SUCCESS in 3m 49s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/15/check/build-openstack-releasenotes/90c1390/html/ : SUCCESS in 3m 12s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/49/652049/15/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/606dbe9/ : SUCCESS in 1h 27m 25s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/49/652049/15/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/53815aa/ : FAILURE in 1h 04m 28s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/15/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/167078b/ : FAILURE in 1h 07m 06s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/49/652049/15/check/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/c460c22/ : FAILURE in 50m 58s","accounts_in_message":[],"_revision_number":15},{"id":"e687225a0aa9d4fdd81e4f1a5fe5dda373da260c","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-06-27 12:39:21.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"0f7041173fec88aa5ca5567658b57b91220be50a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:25:28.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/16/check/openstack-ansible-linters/382e18b/ : SUCCESS in 5m 38s\n- openstack-tox-docs http://logs.openstack.org/49/652049/16/check/openstack-tox-docs/1c99dfe/html/ : SUCCESS in 2m 43s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/16/check/build-openstack-releasenotes/85636c5/html/ : SUCCESS in 3m 12s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/49/652049/16/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/f2df6e1/ : SUCCESS in 1h 40m 15s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/49/652049/16/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/f7e99e2/ : FAILURE in 1h 04m 56s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/16/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/e827f80/ : FAILURE in 1h 10m 37s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/49/652049/16/check/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/62cf6ff/ : SUCCESS in 1h 00m 49s","accounts_in_message":[],"_revision_number":16},{"id":"55c4485635f5f324d889e439b8a756e582ca593d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-28 07:11:25.000000000","message":"Patch Set 17: Commit message was updated.","accounts_in_message":[],"_revision_number":17},{"id":"9d4aa7c4079355be7e69f232d2729d7101c21bb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 08:52:45.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/17/check/openstack-ansible-linters/3b158cb/ : SUCCESS in 5m 35s\n- openstack-tox-docs http://logs.openstack.org/49/652049/17/check/openstack-tox-docs/ca0d367/html/ : SUCCESS in 3m 08s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/17/check/build-openstack-releasenotes/89b7ba2/html/ : SUCCESS in 3m 16s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/49/652049/17/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/87a4f42/ : SUCCESS in 1h 22m 53s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/49/652049/17/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/604e176/ : SUCCESS in 1h 04m 31s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/17/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/2b57300/ : SUCCESS in 1h 00m 19s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/49/652049/17/check/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/1fb454e/ : SUCCESS in 1h 01m 19s","accounts_in_message":[],"_revision_number":17},{"id":"91bbbe761f9672e008dff7b2b369b17b58a1fb75","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-28 10:42:51.000000000","message":"Patch Set 17: Code-Review+2\n\nEven if it\u0027s not the latest standards, it\u0027s fine for me, as it\u0027s an improvement.","accounts_in_message":[],"_revision_number":17},{"id":"e80ea176bab910dd28af38e28f96baceffb7ce61","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-28 12:16:27.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"fac373f93b3feaa725d44ed4bf9c9154c538409c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 12:16:41.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"b12f6fbbf2bafa1c7cb18149f809d76264f4a7a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 14:18:48.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/49/652049/17/gate/openstack-ansible-linters/35c53af/ : SUCCESS in 5m 02s\n- openstack-tox-docs http://logs.openstack.org/49/652049/17/gate/openstack-tox-docs/1900d31/html/ : SUCCESS in 2m 57s\n- build-openstack-releasenotes http://logs.openstack.org/49/652049/17/gate/build-openstack-releasenotes/3ab2615/html/ : SUCCESS in 2m 52s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/49/652049/17/gate/openstack-ansible-deploy-aio_metal_octavia-centos-7/5f926a1/ : SUCCESS in 1h 59m 19s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/49/652049/17/gate/openstack-ansible-deploy-aio_metal_octavia-debian-stable/376a934/ : SUCCESS in 1h 05m 45s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/49/652049/17/gate/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/f058fd2/ : SUCCESS in 1h 00m 14s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/49/652049/17/gate/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/2912161/ : SUCCESS in 1h 08m 56s","accounts_in_message":[],"_revision_number":17},{"id":"958caa0e1cabd58dbdbeb2c40450e71447928e8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 14:18:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"6aa925e792e6dc8f303d99595fb6406d22e95952","revisions":{"8140012b9b52535ce946922b35b2df691aec62d4":{"kind":"REWORK","_number":1,"created":"2019-04-12 11:59:00.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/49/652049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/49/652049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/1"}}},"commit":{"parents":[{"commit":"920cda835c4445aeb43ef0bc8a3e5f21d4450aea","subject":"Update master for stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/920cda835c4445aeb43ef0bc8a3e5f21d4450aea"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 11:36:32.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 11:43:24.000000000","tz":-240},"subject":"Add support for using distribution packages for OpenStack services","message":"Add support for using distribution packages for OpenStack services\n\nDistributions provide packages for the OpenStack services so we add\nsupport for using these instead of the pip ones.\n\nChange-Id: I7eb1cbe2c80ee889d2ae08dcfed6a19cc1bd3415\nImplements: blueprint openstack-distribution-packages\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8140012b9b52535ce946922b35b2df691aec62d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8140012b9b52535ce946922b35b2df691aec62d4"}]},"branch":"refs/heads/master"},"f0d2183d8dc31ed5779103c996e78865b2af1354":{"kind":"REWORK","_number":2,"created":"2019-04-12 12:03:36.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/49/652049/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/49/652049/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/2"}}},"commit":{"parents":[{"commit":"48ab3f8baee8006f9e58c0653eaee364a69739d6","subject":"debian: add support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/48ab3f8baee8006f9e58c0653eaee364a69739d6"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 11:36:32.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 12:03:27.000000000","tz":-240},"subject":"Add support for using distribution packages for OpenStack services","message":"Add support for using distribution packages for OpenStack services\n\nDistributions provide packages for the OpenStack services so we add\nsupport for using these instead of the pip ones.\n\nChange-Id: I7eb1cbe2c80ee889d2ae08dcfed6a19cc1bd3415\nImplements: blueprint openstack-distribution-packages\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f0d2183d8dc31ed5779103c996e78865b2af1354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f0d2183d8dc31ed5779103c996e78865b2af1354"}]},"branch":"refs/heads/master"},"8b01173b0659086ee98bd9ebd762ce2f3980d84b":{"kind":"REWORK","_number":3,"created":"2019-04-12 13:31:05.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/49/652049/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/49/652049/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/3"}}},"commit":{"parents":[{"commit":"48ab3f8baee8006f9e58c0653eaee364a69739d6","subject":"debian: add support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/48ab3f8baee8006f9e58c0653eaee364a69739d6"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 11:36:32.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 13:30:51.000000000","tz":-240},"subject":"Add support for using distribution packages for OpenStack services","message":"Add support for using distribution packages for OpenStack services\n\nDistributions provide packages for the OpenStack services so we add\nsupport for using these instead of the pip ones.\n\nChange-Id: I7eb1cbe2c80ee889d2ae08dcfed6a19cc1bd3415\nImplements: blueprint openstack-distribution-packages\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8b01173b0659086ee98bd9ebd762ce2f3980d84b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8b01173b0659086ee98bd9ebd762ce2f3980d84b"}]},"branch":"refs/heads/master"},"4a2bf594ebd69f7e861639d0d23e5ec91c92c0b8":{"kind":"REWORK","_number":4,"created":"2019-04-12 16:05:02.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/49/652049/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/49/652049/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/4"}}},"commit":{"parents":[{"commit":"48ab3f8baee8006f9e58c0653eaee364a69739d6","subject":"debian: add support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/48ab3f8baee8006f9e58c0653eaee364a69739d6"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 11:36:32.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 15:33:53.000000000","tz":-240},"subject":"Add support for using distribution packages for OpenStack services","message":"Add support for using distribution packages for OpenStack services\n\nDistributions provide packages for the OpenStack services so we add\nsupport for using these instead of the pip ones.\n\nChange-Id: I7eb1cbe2c80ee889d2ae08dcfed6a19cc1bd3415\nImplements: blueprint openstack-distribution-packages\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4a2bf594ebd69f7e861639d0d23e5ec91c92c0b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4a2bf594ebd69f7e861639d0d23e5ec91c92c0b8"}]},"branch":"refs/heads/master"},"71fdd406c385ca586490b3792a04dd5cbfa93f64":{"kind":"REWORK","_number":5,"created":"2019-04-12 19:07:48.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/49/652049/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/49/652049/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/5"}}},"commit":{"parents":[{"commit":"48ab3f8baee8006f9e58c0653eaee364a69739d6","subject":"debian: add support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/48ab3f8baee8006f9e58c0653eaee364a69739d6"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 11:36:32.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 19:07:43.000000000","tz":-240},"subject":"Add support for using distribution packages for OpenStack services","message":"Add support for using distribution packages for OpenStack services\n\nDistributions provide packages for the OpenStack services so we add\nsupport for using these instead of the pip ones.\n\nChange-Id: I7eb1cbe2c80ee889d2ae08dcfed6a19cc1bd3415\nImplements: blueprint openstack-distribution-packages\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/71fdd406c385ca586490b3792a04dd5cbfa93f64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/71fdd406c385ca586490b3792a04dd5cbfa93f64"}]},"branch":"refs/heads/master"},"9d91ab999b45d87d40bbd4561e9d05513f1ce907":{"kind":"REWORK","_number":6,"created":"2019-04-12 21:02:33.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/49/652049/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/49/652049/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/6"}}},"commit":{"parents":[{"commit":"48ab3f8baee8006f9e58c0653eaee364a69739d6","subject":"debian: add support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/48ab3f8baee8006f9e58c0653eaee364a69739d6"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 11:36:32.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 21:02:03.000000000","tz":-240},"subject":"Add support for using distribution packages for OpenStack services","message":"Add support for using distribution packages for OpenStack services\n\nDistributions provide packages for the OpenStack services so we add\nsupport for using these instead of the pip ones.\n\nChange-Id: I7eb1cbe2c80ee889d2ae08dcfed6a19cc1bd3415\nImplements: blueprint openstack-distribution-packages\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9d91ab999b45d87d40bbd4561e9d05513f1ce907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9d91ab999b45d87d40bbd4561e9d05513f1ce907"}]},"branch":"refs/heads/master"},"29d01f97f3d5df3c0b6e8484a63d435b803aefec":{"kind":"REWORK","_number":7,"created":"2019-04-13 01:19:55.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/49/652049/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/49/652049/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/7"}}},"commit":{"parents":[{"commit":"48ab3f8baee8006f9e58c0653eaee364a69739d6","subject":"debian: add support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/48ab3f8baee8006f9e58c0653eaee364a69739d6"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 11:36:32.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-13 01:19:47.000000000","tz":-240},"subject":"Add support for using distribution packages for OpenStack services","message":"Add support for using distribution packages for OpenStack services\n\nDistributions provide packages for the OpenStack services so we add\nsupport for using these instead of the pip ones.\n\nChange-Id: I7eb1cbe2c80ee889d2ae08dcfed6a19cc1bd3415\nImplements: blueprint openstack-distribution-packages\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/29d01f97f3d5df3c0b6e8484a63d435b803aefec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/29d01f97f3d5df3c0b6e8484a63d435b803aefec"}]},"branch":"refs/heads/master"},"5214a834fbeae51e003e73633cc154ac3195bb11":{"kind":"REWORK","_number":8,"created":"2019-04-18 15:44:02.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/49/652049/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/49/652049/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/8"}}},"commit":{"parents":[{"commit":"48ab3f8baee8006f9e58c0653eaee364a69739d6","subject":"debian: add support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/48ab3f8baee8006f9e58c0653eaee364a69739d6"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 11:36:32.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-18 15:43:38.000000000","tz":-240},"subject":"Add support for using distribution packages for OpenStack services","message":"Add support for using distribution packages for OpenStack services\n\nDistributions provide packages for the OpenStack services so we add\nsupport for using these instead of the pip ones.\n\nChange-Id: I7eb1cbe2c80ee889d2ae08dcfed6a19cc1bd3415\nImplements: blueprint openstack-distribution-packages\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/5214a834fbeae51e003e73633cc154ac3195bb11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/5214a834fbeae51e003e73633cc154ac3195bb11"}]},"branch":"refs/heads/master"},"b2ebed47dbeb6bb4fc12288d427d0987d6bf8fce":{"kind":"REWORK","_number":9,"created":"2019-04-20 12:07:06.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/49/652049/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/49/652049/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/9"}}},"commit":{"parents":[{"commit":"48ab3f8baee8006f9e58c0653eaee364a69739d6","subject":"debian: add support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/48ab3f8baee8006f9e58c0653eaee364a69739d6"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 11:36:32.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-20 11:58:02.000000000","tz":-240},"subject":"Add support for using distribution packages for OpenStack services","message":"Add support for using distribution packages for OpenStack services\n\nDistributions provide packages for the OpenStack services so we add\nsupport for using these instead of the pip ones.\n\nChange-Id: I7eb1cbe2c80ee889d2ae08dcfed6a19cc1bd3415\nDepend-On: https://review.opendev.org/653849\nImplements: blueprint openstack-distribution-packages\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b2ebed47dbeb6bb4fc12288d427d0987d6bf8fce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b2ebed47dbeb6bb4fc12288d427d0987d6bf8fce"}]},"branch":"refs/heads/master"},"06d36f14481dd2b5f5807999d6ca626bcd56c60d":{"kind":"REWORK","_number":10,"created":"2019-04-23 13:46:44.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/49/652049/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/49/652049/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/10"}}},"commit":{"parents":[{"commit":"d4108826eb18c3bd7a45a736aee38a556afa5c80","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/d4108826eb18c3bd7a45a736aee38a556afa5c80"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 11:36:32.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-23 13:46:29.000000000","tz":-240},"subject":"Add support for using distribution packages for OpenStack services","message":"Add support for using distribution packages for OpenStack services\n\nDistributions provide packages for the OpenStack services so we add\nsupport for using these instead of the pip ones.\n\nChange-Id: I7eb1cbe2c80ee889d2ae08dcfed6a19cc1bd3415\nDepend-On: https://review.opendev.org/653849\nImplements: blueprint openstack-distribution-packages\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/06d36f14481dd2b5f5807999d6ca626bcd56c60d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/06d36f14481dd2b5f5807999d6ca626bcd56c60d"}]},"branch":"refs/heads/master"},"f28b35907f4e953d22df4f915e3883acff0dd115":{"kind":"REWORK","_number":11,"created":"2019-05-06 17:42:54.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/49/652049/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/49/652049/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/11"}}},"commit":{"parents":[{"commit":"9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 11:36:32.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-05-06 17:42:39.000000000","tz":-240},"subject":"Add support for using distribution packages for OpenStack services","message":"Add support for using distribution packages for OpenStack services\n\nDistributions provide packages for the OpenStack services so we add\nsupport for using these instead of the pip ones.\n\nChange-Id: I7eb1cbe2c80ee889d2ae08dcfed6a19cc1bd3415\nDepend-On: https://review.opendev.org/653849\nImplements: blueprint openstack-distribution-packages\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f28b35907f4e953d22df4f915e3883acff0dd115"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f28b35907f4e953d22df4f915e3883acff0dd115"}]},"branch":"refs/heads/master"},"163051f4e2c8f56addb8805d1862d98070f67a92":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2019-05-07 12:59:59.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/49/652049/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/49/652049/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/12"}}},"commit":{"parents":[{"commit":"9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 11:36:32.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-05-07 12:58:57.000000000","tz":-240},"subject":"Add support for using distribution packages for OpenStack services","message":"Add support for using distribution packages for OpenStack services\n\nDistributions provide packages for the OpenStack services so we add\nsupport for using these instead of the pip ones.\n\nChange-Id: I7eb1cbe2c80ee889d2ae08dcfed6a19cc1bd3415\nDepends-On: https://review.opendev.org/653849\nImplements: blueprint openstack-distribution-packages\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/163051f4e2c8f56addb8805d1862d98070f67a92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/163051f4e2c8f56addb8805d1862d98070f67a92"}]},"branch":"refs/heads/master"},"ed2aac8ce1d48208dafa447cd9a45c86944476fb":{"kind":"REWORK","_number":13,"created":"2019-05-07 14:16:05.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/49/652049/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/49/652049/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/13"}}},"commit":{"parents":[{"commit":"9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 11:36:32.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-05-07 14:13:45.000000000","tz":-240},"subject":"Add support for using distribution packages for OpenStack services","message":"Add support for using distribution packages for OpenStack services\n\nDistributions provide packages for the OpenStack services so we add\nsupport for using these instead of the pip ones.\n\nChange-Id: I7eb1cbe2c80ee889d2ae08dcfed6a19cc1bd3415\nDepends-On: https://review.opendev.org/653849\nImplements: blueprint openstack-distribution-packages\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ed2aac8ce1d48208dafa447cd9a45c86944476fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ed2aac8ce1d48208dafa447cd9a45c86944476fb"}]},"branch":"refs/heads/master"},"2c76800c912f1f152fcb7def994af0268b5fa730":{"kind":"REWORK","_number":14,"created":"2019-05-08 17:15:40.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/49/652049/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/49/652049/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/14"}}},"commit":{"parents":[{"commit":"9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 11:36:32.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-05-08 17:15:25.000000000","tz":-240},"subject":"Add support for using distribution packages for OpenStack services","message":"Add support for using distribution packages for OpenStack services\n\nDistributions provide packages for the OpenStack services so we add\nsupport for using these instead of the pip ones.\n\nChange-Id: I7eb1cbe2c80ee889d2ae08dcfed6a19cc1bd3415\nDepends-On: https://review.opendev.org/653849\nImplements: blueprint openstack-distribution-packages\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/2c76800c912f1f152fcb7def994af0268b5fa730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/2c76800c912f1f152fcb7def994af0268b5fa730"}]},"branch":"refs/heads/master"},"be66b69ff0bb2e81b8643d5308cdcabe11cac7d9":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2019-06-27 08:55:06.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/49/652049/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/49/652049/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/15"}}},"commit":{"parents":[{"commit":"9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 11:36:32.000000000","tz":-240},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-27 08:55:03.000000000","tz":0},"subject":"Add support for using distribution packages for OpenStack services","message":"Add support for using distribution packages for OpenStack services\n\nDistributions provide packages for the OpenStack services so we add\nsupport for using these instead of the pip ones.\n\nChange-Id: I7eb1cbe2c80ee889d2ae08dcfed6a19cc1bd3415\nDepends-On: Ia5fda5d417b79189d048c8891b84d57331df1404\nImplements: blueprint openstack-distribution-packages\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/be66b69ff0bb2e81b8643d5308cdcabe11cac7d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/be66b69ff0bb2e81b8643d5308cdcabe11cac7d9"}]},"branch":"refs/heads/master"},"3a6ccfb344dc14aa2648f5ce22d164280b296e8f":{"kind":"REWORK","_number":16,"created":"2019-06-27 12:39:21.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/49/652049/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/49/652049/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/16"}}},"commit":{"parents":[{"commit":"9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 11:36:32.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-06-27 12:38:36.000000000","tz":-240},"subject":"Add support for using distribution packages for OpenStack services","message":"Add support for using distribution packages for OpenStack services\n\nDistributions provide packages for the OpenStack services so we add\nsupport for using these instead of the pip ones.\n\nChange-Id: I7eb1cbe2c80ee889d2ae08dcfed6a19cc1bd3415\nDepends-On: Ia5fda5d417b79189d048c8891b84d57331df1404\nImplements: blueprint openstack-distribution-packages\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3a6ccfb344dc14aa2648f5ce22d164280b296e8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3a6ccfb344dc14aa2648f5ce22d164280b296e8f"}]},"branch":"refs/heads/master"},"6aa925e792e6dc8f303d99595fb6406d22e95952":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2019-06-28 07:11:25.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/49/652049/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/49/652049/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/49/652049/17"}}},"commit":{"parents":[{"commit":"9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-04-12 11:36:32.000000000","tz":-240},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-28 07:11:21.000000000","tz":0},"subject":"Add support for using distribution packages for OpenStack services","message":"Add support for using distribution packages for OpenStack services\n\nDistributions provide packages for the OpenStack services so we add\nsupport for using these instead of the pip ones.\n\nChange-Id: I7eb1cbe2c80ee889d2ae08dcfed6a19cc1bd3415\nDepends-On: Ide70b5d8f67d8c8a87e3f16671f0f7fb72338b89\nDepends-On: I8de48eb1fb4c8d321098ca54b9e21270edc7ac87\nDepends-On: Ia5fda5d417b79189d048c8891b84d57331df1404\nImplements: blueprint openstack-distribution-packages\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/6aa925e792e6dc8f303d99595fb6406d22e95952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/6aa925e792e6dc8f303d99595fb6406d22e95952"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
