)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~657266","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~Icf8eed973a4ab5c93b57423eda12b9b12a3e3269","project":"openstack/openstack-ansible-os_octavia","branch":"master","hashtags":[],"change_id":"Icf8eed973a4ab5c93b57423eda12b9b12a3e3269","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"ABANDONED","created":"2019-05-05 13:11:44.000000000","updated":"2019-07-17 08:59:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5249e9a6131789871c742a24b113045003018dd3","_number":657266,"virtual_id_number":657266,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"date":"2019-07-10 15:15:59.000000000","_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-05-06 12:49:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":1,"date":"2019-05-20 20:43:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2019-05-21 10:22:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-06 12:49:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-20 20:43:13.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-21 10:22:15.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"}],"messages":[{"id":"bf5f88bf012a2bc40e532141e85f13ee8d1d6783","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-05 13:11:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1777d68c8ac7118a7ef530360e9ae6b98174da6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 14:45:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/66/657266/1/check/openstack-ansible-linters/9bbe7f8/ : SUCCESS in 4m 14s\n- openstack-tox-docs http://logs.openstack.org/66/657266/1/check/openstack-tox-docs/98f48f8/html/ : SUCCESS in 2m 17s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/66/657266/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/8f155b3/ : SUCCESS in 1h 14m 58s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/66/657266/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/452fcbc/ : SUCCESS in 53m 50s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/66/657266/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/59b07ff/ : SUCCESS in 1h 00m 08s","accounts_in_message":[],"_revision_number":1},{"id":"73f46f4dea3bebddf20404391cae39698fa1a5a3","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-06 11:24:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fb24035bbc920ea868a871a8aeefb032df3700ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 12:49:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/66/657266/2/check/openstack-ansible-linters/748ff0d/ : SUCCESS in 5m 48s\n- openstack-tox-docs http://logs.openstack.org/66/657266/2/check/openstack-tox-docs/fdf9ea7/html/ : SUCCESS in 2m 32s\n- build-openstack-releasenotes http://logs.openstack.org/66/657266/2/check/build-openstack-releasenotes/3c87552/html/ : SUCCESS in 2m 08s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/66/657266/2/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/554da40/ : SUCCESS in 1h 19m 36s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/66/657266/2/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/fece068/ : SUCCESS in 1h 02m 47s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/66/657266/2/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/bccbf98/ : SUCCESS in 58m 26s","accounts_in_message":[],"_revision_number":2},{"id":"ddd3bf6350bae9dbd461aac172ec9318efb26102","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-20 20:43:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cee1ab6811a9030df1189083c6a6a42ba1326d3f","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 10:22:15.000000000","message":"Patch Set 2: Code-Review+1\n\nSee url in https://review.opendev.org/#/c/657202/2/tox.ini","accounts_in_message":[],"_revision_number":2},{"id":"d61a676ac40e15be2891e62980173769671bda6b","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-07-10 15:15:59.000000000","message":"Patch Set 2:\n\nWhen I update the patch,there has a merge confict.  So please  review this patch,  https://review.opendev.org/#/c/670063/","accounts_in_message":[],"_revision_number":2},{"id":"5249e9a6131789871c742a24b113045003018dd3","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-07-17 08:59:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"69fee30c6a431b8a04ca758dfcb2ef8859c5361d","revisions":{"3355ef25da7ab69c2d389d9a877c4c5de9292e7a":{"kind":"REWORK","_number":1,"created":"2019-05-05 13:11:44.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/66/657266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/66/657266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/66/657266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/66/657266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/66/657266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/66/657266/1"}}},"commit":{"parents":[{"commit":"9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 13:11:17.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 13:11:42.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Icf8eed973a4ab5c93b57423eda12b9b12a3e3269\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3355ef25da7ab69c2d389d9a877c4c5de9292e7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3355ef25da7ab69c2d389d9a877c4c5de9292e7a"}]},"branch":"refs/heads/master"},"69fee30c6a431b8a04ca758dfcb2ef8859c5361d":{"kind":"REWORK","_number":2,"created":"2019-05-06 11:24:48.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/66/657266/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/66/657266/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/66/657266/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/66/657266/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/66/657266/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/66/657266/2"}}},"commit":{"parents":[{"commit":"9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 13:11:17.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-06 11:24:27.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Icf8eed973a4ab5c93b57423eda12b9b12a3e3269\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/69fee30c6a431b8a04ca758dfcb2ef8859c5361d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/69fee30c6a431b8a04ca758dfcb2ef8859c5361d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
