)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~657662","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I0906c7cf80c3345fb993a71fe190fa2f6baa55e4","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"bug/1842117","hashtags":[],"change_id":"I0906c7cf80c3345fb993a71fe190fa2f6baa55e4","subject":"Octavia certificate distribution","status":"MERGED","created":"2019-05-07 18:38:44.000000000","updated":"2019-09-26 22:59:19.000000000","submitted":"2019-09-26 22:58:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657662-1569538682824-e03b07bf","meta_rev_id":"258fb27492e6466c0b02b7c4600ea09f29e5f651","_number":657662,"virtual_id_number":657662,"owner":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"date":"2019-09-20 15:36:38.000000000","_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-26 22:58:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-09-20 21:55:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":2,"date":"2019-09-26 20:35:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":1,"date":"2019-09-26 20:35:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-07 18:40:11.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2019-05-07 22:31:12.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-08-07 09:49:25.000000000","updated_by":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"reviewer":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"state":"REVIEWER"},{"updated":"2019-09-20 21:55:08.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-09-26 20:35:14.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-09-26 22:58:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"833a047caeaed380f7dfc70f67356d194fc81762","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-05-07 18:38:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07235b31a2f3c26b971a6102fdd324196c62f75f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 20:06:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/62/657662/1/check/openstack-ansible-linters/e8d46a8/ : SUCCESS in 5m 34s\n- openstack-tox-docs http://logs.openstack.org/62/657662/1/check/openstack-tox-docs/445030f/html/ : SUCCESS in 2m 22s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/62/657662/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/337af06/ : SUCCESS in 1h 19m 11s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/62/657662/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/77c0466/ : SUCCESS in 1h 07m 13s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/62/657662/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/93379f9/ : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7f86a178e92d5ef6b6c5505d7e7f979ea4975729","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-05-07 22:31:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"714d2687de9fdc54d275d4c660e258812cd1a3aa","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-28 12:21:36.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"828767b2f2f01fcbfa6e4c559d643916f2ac9f76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 13:42:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/62/657662/2/check/openstack-ansible-linters/74d2a08/ : SUCCESS in 5m 59s\n- openstack-tox-docs http://logs.openstack.org/62/657662/2/check/openstack-tox-docs/85d8fc7/html/ : SUCCESS in 3m 07s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/62/657662/2/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/cdd987d/ : FAILURE in 1h 15m 02s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/62/657662/2/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/71a62bc/ : FAILURE in 50m 59s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/62/657662/2/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/d4433a3/ : FAILURE in 49m 07s","accounts_in_message":[],"_revision_number":2},{"id":"8d8b1cd6729345344377892299f60e3c544baa47","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-06-28 14:35:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"762bcbbaed5f15ae369dd588f82c09db86aff7c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 15:53:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/62/657662/2/check/openstack-ansible-linters/37b2c90/ : SUCCESS in 5m 22s\n- openstack-tox-docs http://logs.openstack.org/62/657662/2/check/openstack-tox-docs/d70a3ae/html/ : SUCCESS in 2m 54s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/62/657662/2/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/0a5384d/ : FAILURE in 1h 11m 02s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/62/657662/2/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/340aa18/ : FAILURE in 54m 04s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/62/657662/2/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/2132bcc/ : FAILURE in 51m 47s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/62/657662/2/check/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/08182a8/ : FAILURE in 46m 02s","accounts_in_message":[],"_revision_number":2},{"id":"79b5e2583dcadce3590bedd89faf57094e51140d","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-06-28 18:07:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"76024a61a6d7d6402f645aa3302d33435b616e85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 19:24:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/62/657662/3/check/openstack-ansible-linters/32f4ad6/ : SUCCESS in 5m 46s\n- openstack-tox-docs http://logs.openstack.org/62/657662/3/check/openstack-tox-docs/133ed05/html/ : SUCCESS in 3m 20s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/62/657662/3/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/1566caa/ : FAILURE in 1h 10m 54s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/62/657662/3/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/9ca45fa/ : FAILURE in 57m 54s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/62/657662/3/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/60d8df9/ : FAILURE in 53m 03s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/62/657662/3/check/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/e147667/ : FAILURE in 49m 28s","accounts_in_message":[],"_revision_number":3},{"id":"6a24011d2aebc9d3f62624a94a37ec1b05a2fd8c","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-06-28 19:49:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9e69c239e75516ccee62f84deef6bd7cfae306d8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-28 20:05:45.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"112e26f0d37ad610fbb202d3131d61c73347b580","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-06-28 21:19:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f94f52061fa2c6185b906f736dee5752887a4d8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 23:13:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/62/657662/5/check/openstack-ansible-linters/315fea3/ : SUCCESS in 8m 06s\n- openstack-tox-docs http://logs.openstack.org/62/657662/5/check/openstack-tox-docs/4aaa899/html/ : SUCCESS in 2m 35s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/62/657662/5/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/d06c516/ : SUCCESS in 1h 52m 59s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/62/657662/5/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/0f8a6ab/ : SUCCESS in 1h 23m 47s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/62/657662/5/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/9b06bb1/ : SUCCESS in 1h 03m 03s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/62/657662/5/check/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/b687c0a/ : SUCCESS in 1h 16m 57s","accounts_in_message":[],"_revision_number":5},{"id":"ea7b3eaf170de7b5678b4b0c1c650fa4616cc999","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-07-01 12:19:36.000000000","message":"Patch Set 5:\n\n\u003e (1 comment)\n\nActually we are both wrong. Do you mind to review it again, please?","accounts_in_message":[],"_revision_number":5},{"id":"8ba082d094c1f2740e95ec52d37b7cdcf7461ba5","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-07-10 13:26:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"09611c8543decaa68d9240173b0862aad1a33a41","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-07-23 19:06:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0fc6491ce61811be2106fac92bd8cfede95e7d92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 20:56:19.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/62/657662/6/check/openstack-ansible-linters/7cbdfc7/ : FAILURE in 10m 11s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/62/657662/6/check/openstack-ansible-deploy-aio_metal-debian-stable/8a98704/ : SUCCESS in 1h 07m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/62/657662/6/check/openstack-ansible-deploy-aio_metal-centos-7/fcc3d05/ : SUCCESS in 1h 42m 33s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/62/657662/6/check/openstack-ansible-deploy-aio_metal-opensuse-15/5b7139d/ : RETRY_LIMIT in 3m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/62/657662/6/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/d79bd54/ : SUCCESS in 1h 15m 26s\n- openstack-tox-docs http://logs.openstack.org/62/657662/6/check/openstack-tox-docs/9ab0c11/html/ : SUCCESS in 3m 14s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/62/657662/6/check/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/ac0c23c/ : SUCCESS in 1h 04m 57s","accounts_in_message":[],"_revision_number":6},{"id":"690f4e399185f46ea94488ea1596673d1fc6258f","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-07-23 22:04:31.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"23cc61c315349844aa4e56e4f8db9d1063374ced","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 23:43:32.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/62/657662/7/check/openstack-ansible-linters/6d22a05/ : SUCCESS in 7m 02s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/62/657662/7/check/openstack-ansible-deploy-aio_metal-debian-stable/84401f3/ : SUCCESS in 1h 23m 26s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/62/657662/7/check/openstack-ansible-deploy-aio_metal-centos-7/5224a88/ : SUCCESS in 1h 35m 25s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/62/657662/7/check/openstack-ansible-deploy-aio_metal-opensuse-15/2c2d353/ : RETRY_LIMIT in 2m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/62/657662/7/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/af4a4b8/ : SUCCESS in 1h 18m 36s\n- openstack-tox-docs http://logs.openstack.org/62/657662/7/check/openstack-tox-docs/13204f5/html/ : SUCCESS in 3m 03s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/62/657662/7/check/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/b5e3a76/ : SUCCESS in 1h 06m 24s","accounts_in_message":[],"_revision_number":7},{"id":"7b38e54ab9119fedeec0024ca37d20ec7c202123","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-07-25 00:14:43.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"14480ac7fd098ed95520e597991b40ee5211e775","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 01:41:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/62/657662/7/check/openstack-ansible-linters/443e87e/ : SUCCESS in 5m 52s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/62/657662/7/check/openstack-ansible-deploy-aio_metal-debian-stable/4f81fd5/ : SUCCESS in 1h 05m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/62/657662/7/check/openstack-ansible-deploy-aio_metal-centos-7/bbd80a0/ : SUCCESS in 1h 25m 01s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/62/657662/7/check/openstack-ansible-deploy-aio_metal-opensuse-15/a4aaae1/ : SUCCESS in 1h 14m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/62/657662/7/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/00099b4/ : SUCCESS in 1h 09m 13s\n- openstack-tox-docs http://logs.openstack.org/62/657662/7/check/openstack-tox-docs/0bfdd16/html/ : SUCCESS in 2m 41s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/62/657662/7/check/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/cfbb41e/ : SUCCESS in 1h 04m 17s","accounts_in_message":[],"_revision_number":7},{"id":"f980025f8d04325f5040274081eae39da7bbf880","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-07-26 13:55:31.000000000","message":"Patch Set 7:\n\nReview please.","accounts_in_message":[],"_revision_number":7},{"id":"40f3ae22d7f0cd06f2c21f9d3da06c12ce5d0cce","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-08-07 09:49:25.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"fd8a2e0f042edb73971a3eb33e1b930f9d8bf8d9","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-08-30 17:21:36.000000000","message":"Abandoned\n\n:(","accounts_in_message":[],"_revision_number":7},{"id":"aa2a34bd2d4a8ec56b21951f0b8e379ad51fc454","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-02 07:51:14.000000000","message":"Patch Set 7:\n\n\u003e Abandoned\n \u003e \n \u003e :(\n\nSorry you\u0027ve abandoned this - i\u0027m just really not understanding how the changes relate to the commit message, the tasks you have moved in octavia_certs.yml are still delegated to the cert setup host?\n\nDoes the commit message need to be improved? Describe the failure you saw and why it happend, what change you have made and what the new behaviour is. This would make the patch easier to review.","accounts_in_message":[],"_revision_number":7},{"id":"8342b745911f3901356692c6d3fdab4b35a9dc9c","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-09-03 16:38:17.000000000","message":"Restored\n\nI will revise the patch.","accounts_in_message":[],"_revision_number":7},{"id":"4e4cb8053c93edf3e63677aa031f823d90931f59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 16:38:29.000000000","message":"Patch Set 7: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"136a00f59bb319202e3db48893346716ee6f02cd","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-09-03 16:38:32.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f9437f5c7d713cb0162fe35170c781e1ad9a02fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 18:39:06.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8b99fba329bb46889791e57a892000df : SUCCESS in 7m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/993cfe0af73947c98b791d095a829ce8 : SUCCESS in 1h 19m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2fe1b83961a844fd98c527a909275441 : SUCCESS in 1h 55m 52s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/81ed9fd21165481593db6c5628f764da : FAILURE in 42m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/470336e946ba47099b397fe2392db26c : SUCCESS in 1h 08m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6f64056e0784ec48667ecfb533e885a : SUCCESS in 4m 33s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/eb8fbf0bb216493fbd86ac58803ee79f : SUCCESS in 1h 41m 40s","accounts_in_message":[],"_revision_number":8},{"id":"0ca4541417cbbdb6aa70ae8f08deaff8812ed402","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-03 21:22:45.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"1949c041487553e9c862e71ff95423109fbc828f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 23:39:38.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d17231f5b5e340949a7a9934ca62e2c5 : SUCCESS in 7m 37s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/1107dd77286746b7b5c89afe94c1f34d : SUCCESS in 1h 19m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3ec5645543464aadb37eff0dfa138c2b : SUCCESS in 2h 14m 49s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1032f03bbd2e47d184ebeb902bc7e2f3 : FAILURE in 42m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/889eee8e8e834adc8586f7e539075446 : SUCCESS in 1h 27m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21e8e1f4b14b41b9b2f8a40f0cf30650 : SUCCESS in 4m 10s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/bccfa35c5a274e86bbfb56e27beda2eb : SUCCESS in 1h 07m 50s","accounts_in_message":[],"_revision_number":8},{"id":"4e93d6dbd013f6097fdad4525c67743094cf6605","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-09-04 19:18:44.000000000","message":"Patch Set 8:\n\nRecheck","accounts_in_message":[],"_revision_number":8},{"id":"b21542127d40846189e25f4b4d535804fa785e78","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-09-04 19:19:28.000000000","message":"Patch Set 8:\n\n\u003e \u003e Abandoned\n \u003e \u003e\n \u003e \u003e :(\n \u003e \n \u003e Sorry you\u0027ve abandoned this - i\u0027m just really not understanding how\n \u003e the changes relate to the commit message, the tasks you have moved\n \u003e in octavia_certs.yml are still delegated to the cert setup host?\n \u003e \n \u003e Does the commit message need to be improved? Describe the failure\n \u003e you saw and why it happend, what change you have made and what the\n \u003e new behaviour is. This would make the patch easier to review.\n\nCould you review it again, please?","accounts_in_message":[],"_revision_number":8},{"id":"2180cbb4e2411c1db2e4b601284b8a55937ca28b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 21:18:16.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/28e854e7ff434722a9b76a2ca4e11313 : SUCCESS in 7m 23s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/0302d2ea996b4a49a0fb881864deeb03 : SUCCESS in 1h 19m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a8fc3fcab7ee4b7180fced828714993b : SUCCESS in 1h 52m 12s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/103315e2451e4d32aa5923bec87f72f8 : FAILURE in 53m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f97a90c40e94635a8408b087087a9ce : SUCCESS in 1h 10m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f54ec16320b2498c9fbf87de9a4d7a91 : SUCCESS in 3m 28s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/0913b86113004214af7f2e91fe0e62a9 : SUCCESS in 1h 27m 16s","accounts_in_message":[],"_revision_number":8},{"id":"39ee27f5aec73cb0335f2ea5e9d550db67961efc","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-09-05 14:04:47.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"f899971607fb9f7777d71bc3b642a600a17e23f6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-05 14:24:47.000000000","message":"Patch Set 8:\n\nrecheck - suse failure was just fixed with https://review.opendev.org/#/c/679904/","accounts_in_message":[],"_revision_number":8},{"id":"63d9ec92b69a4ae71bb7efd736ea8a23bf6adf2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 18:32:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0390c35b677240a5a1e125cdf4e69559 : SUCCESS in 6m 53s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/a658dae4650d4098a0ebd4bfedef3e1f : SUCCESS in 1h 27m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/19340cf737254c0f9e4dc374c48de4ac : SUCCESS in 2h 04m 40s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3d763d45ca104729bc97a02cc690d617 : SUCCESS in 2h 19m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e9de2d54f7745578a0ca74a3a49db48 : SUCCESS in 1h 16m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8d7ba97dd664cd88b977bf0ac153d18 : SUCCESS in 7m 04s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/5ba5c6c23a9c452080cc78e27c66249e : SUCCESS in 1h 17m 20s","accounts_in_message":[],"_revision_number":8},{"id":"33edd755d9a4f23d02783e8d0f26ba74b8b45852","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-20 13:56:02.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"478643985e53c718a5874eeae6650c3dcb12ed41","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-20 13:59:28.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"8c8cd69a40c1be1927cdbd36289f267d3367d069","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-09-20 15:36:38.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"01498c50c3bcd10ec358829b7d5d20d94379898c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 16:17:28.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/57bd2b9484b446038a9a14d13cfc1f20 : SUCCESS in 7m 06s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/f55cc7462d2b4fd8a0f707387b16ce5d : SUCCESS in 1h 34m 23s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4133ceda45c44c02b78a79e89d40969c : SUCCESS in 2h 08m 25s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6e0b0fee757c43f28e1be888a36c1de7 : SUCCESS in 1h 47m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ffc190c1613943ba9a361251e6c977bb : SUCCESS in 1h 32m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e16c16333ab4cd2a7b04129af809708 : SUCCESS in 3m 42s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/c93c3760d0b64c92a2ed0bec9a3e60a3 : FAILURE in 1h 18m 09s","accounts_in_message":[],"_revision_number":9},{"id":"3ce38223694ed3606a360832b56599eab32a6c5c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-20 16:51:05.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"8a8ce93feb1fbf92e86943fe4a8fe9e47ebb69ba","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-20 17:14:08.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6d204fe912ba92fc0e43f6b2d7fa138b6ce1c9a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 19:41:18.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/df2be7688b2c4f8a86f9fd6d1a2e02bc : SUCCESS in 7m 50s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/a7fe92a10e724625936bf3da6ce6c9cf : SUCCESS in 1h 31m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c6008c254e924e108f99ffca2d7272c9 : SUCCESS in 2h 08m 46s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/130508f4bb724993bad8ae18d15abd4c : SUCCESS in 1h 44m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec94cde4741b45c084667e6be38cb6b0 : SUCCESS in 1h 21m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5df799a0b9d04b8e990fd81e5ff8f6ca : SUCCESS in 5m 00s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/8b032ba11ecb413b9fdf7b317cfcb7a4 : SUCCESS in 1h 32m 54s","accounts_in_message":[],"_revision_number":9},{"id":"43516d3f96d824f44fe32ed4f6d913fc3125b831","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-20 21:55:08.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"3724d3fc55b24663e68bc3a6a2c3daf642ddbd66","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-26 20:35:14.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"f17ba1fd038a35548cb368056ace3d737fa204cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 20:35:26.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"9626596f44abfeed26de6ecb68c7099a54c52da9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 22:58:02.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5738d4a12a2a440c8aecc03118b0f236 : SUCCESS in 8m 51s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/0e914be7b2ba4201a4b6d2011643f540 : SUCCESS in 1h 56m 40s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/845923e733cb479b8a150ff7a6957518 : SUCCESS in 2h 17m 25s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e72e781a325840589e54b41975a8732e : SUCCESS in 1h 37m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9686980e0ff54b9cb1a9e8c237d1023d : SUCCESS in 1h 26m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6a8c379fe6d4f69a2514bc3c7ec3b80 : SUCCESS in 3m 22s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/60c210aa0ca94773b50746f46f8d62f5 : SUCCESS in 1h 41m 16s","accounts_in_message":[],"_revision_number":9},{"id":"1e09676a102193e5a271d2dbee0abbe9c9a6e8ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 22:58:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"389c5691068644318a19e8c9c1a11fed42355879","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 22:59:19.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02e3c657e3bc4fd1960a3387860c895e : SUCCESS in 59s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"833b58da2c71f0e6e77253388a64d00d1ccf83f3","revisions":{"476ca6ba7017bdb1dd3a3ec65bc0812399617419":{"kind":"REWORK","_number":1,"created":"2019-05-07 18:38:44.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/62/657662/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/62/657662/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/1"}}},"commit":{"parents":[{"commit":"9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-05-07 18:22:08.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-05-07 18:34:50.000000000","tz":-240},"subject":"Updated certificate distribution task","message":"Updated certificate distribution task\n\nReplaced post_install tasks that worked only if setup host is\ndeployment host or localhost\n\nAdditionally moved CA certificate generation under\noctavia_generate_certs condition\n\nChange-Id: I0906c7cf80c3345fb993a71fe190fa2f6baa55e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/476ca6ba7017bdb1dd3a3ec65bc0812399617419"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/476ca6ba7017bdb1dd3a3ec65bc0812399617419"}]},"branch":"refs/heads/master"},"e13c66f5233502044bdcb1d74804bc7b8a5acebc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-28 12:21:36.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/62/657662/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/62/657662/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/2"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-05-07 18:22:08.000000000","tz":-240},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-28 12:21:36.000000000","tz":0},"subject":"Updated certificate distribution task","message":"Updated certificate distribution task\n\nReplaced post_install tasks that worked only if setup host is\ndeployment host or localhost\n\nAdditionally moved CA certificate generation under\noctavia_generate_certs condition\n\nChange-Id: I0906c7cf80c3345fb993a71fe190fa2f6baa55e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e13c66f5233502044bdcb1d74804bc7b8a5acebc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e13c66f5233502044bdcb1d74804bc7b8a5acebc"}]},"branch":"refs/heads/master"},"efe5842d4a408391a1afbb6faf2250c96cfa5077":{"kind":"REWORK","_number":3,"created":"2019-06-28 18:07:50.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/62/657662/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/62/657662/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/3"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-05-07 18:22:08.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-06-28 18:07:07.000000000","tz":-240},"subject":"Updated certificate distribution task","message":"Updated certificate distribution task\n\nReplaced post_install tasks that worked only if setup host is\ndeployment host or localhost\n\nAdditionally moved CA certificate generation under\noctavia_generate_certs condition\n\nChange-Id: I0906c7cf80c3345fb993a71fe190fa2f6baa55e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/efe5842d4a408391a1afbb6faf2250c96cfa5077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/efe5842d4a408391a1afbb6faf2250c96cfa5077"}]},"branch":"refs/heads/master"},"1880497c4dbd931ff600cd2297f5790cad4f8d91":{"kind":"REWORK","_number":4,"created":"2019-06-28 19:49:24.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/62/657662/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/62/657662/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/4"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-05-07 18:22:08.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-06-28 19:49:03.000000000","tz":-240},"subject":"Updated certificate distribution task","message":"Updated certificate distribution task\n\nReplaced post_install tasks that worked only if setup host is\ndeployment host or localhost\n\nAdditionally moved CA certificate generation under\noctavia_generate_certs condition\n\nChange-Id: I0906c7cf80c3345fb993a71fe190fa2f6baa55e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1880497c4dbd931ff600cd2297f5790cad4f8d91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1880497c4dbd931ff600cd2297f5790cad4f8d91"}]},"branch":"refs/heads/master"},"f23dcdaeef029bd684b802e28857827b996110e4":{"kind":"REWORK","_number":5,"created":"2019-06-28 21:19:13.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/62/657662/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/62/657662/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/5"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-05-07 18:22:08.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-06-28 21:18:47.000000000","tz":-240},"subject":"Updated certificate distribution task","message":"Updated certificate distribution task\n\nReplaced post_install tasks that worked only if setup host is\ndeployment host or localhost\n\nAdditionally moved CA certificate generation under\noctavia_generate_certs condition\n\nChange-Id: I0906c7cf80c3345fb993a71fe190fa2f6baa55e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f23dcdaeef029bd684b802e28857827b996110e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f23dcdaeef029bd684b802e28857827b996110e4"}]},"branch":"refs/heads/master"},"d2b04550333d44319804a23fa4e4c4bb6b5e3b3a":{"kind":"REWORK","_number":6,"created":"2019-07-23 19:06:32.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/62/657662/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/62/657662/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/6"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-05-07 18:22:08.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-07-23 19:06:10.000000000","tz":-240},"subject":"Updated certificate distribution task","message":"Updated certificate distribution task\n\nReplaced post_install tasks that worked only if setup host is\ndeployment host or localhost\n\nAdditionally moved CA certificate generation under\noctavia_generate_certs condition\n\nChange-Id: I0906c7cf80c3345fb993a71fe190fa2f6baa55e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/d2b04550333d44319804a23fa4e4c4bb6b5e3b3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/d2b04550333d44319804a23fa4e4c4bb6b5e3b3a"}]},"branch":"refs/heads/master"},"1361f6ed97a2115878de75d37f48540badf2643e":{"kind":"REWORK","_number":7,"created":"2019-07-23 22:04:31.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/62/657662/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/62/657662/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/7"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-05-07 18:22:08.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-07-23 22:04:21.000000000","tz":-240},"subject":"Updated certificate distribution task","message":"Updated certificate distribution task\n\nReplaced post_install tasks that worked only if setup host is\ndeployment host or localhost\n\nAdditionally moved CA certificate generation under\noctavia_generate_certs condition\n\nChange-Id: I0906c7cf80c3345fb993a71fe190fa2f6baa55e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1361f6ed97a2115878de75d37f48540badf2643e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1361f6ed97a2115878de75d37f48540badf2643e"}]},"branch":"refs/heads/master"},"13cc76cb8e2a67e5ac2209bd55f3f56b5d87dc73":{"kind":"REWORK","_number":8,"created":"2019-09-03 16:38:32.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/62/657662/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/62/657662/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/8"}}},"commit":{"parents":[{"commit":"b28287a0fbaf7a88c343ee98fe6a2bc5e275e7a5","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b28287a0fbaf7a88c343ee98fe6a2bc5e275e7a5"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-09-03 16:05:58.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-09-03 16:34:40.000000000","tz":-240},"subject":"Octavia certificate distribution","message":"Octavia certificate distribution\n\nCurrent copy certificates task works only if octavia_cert_setup_host is\nlocalhost, because it copies from local path. See bug 1842117\n\nTo fix this issue certs are slurped and registered after they are\ncreated.\nIf octavia_cert_setup_host is \u0027localhost\u0027 cert variables are\nregistered to task_vars scope.\nIf octavia_cert_setup_host is one of the octavia containers or utility\ncontainers, i.e. setup host exists in inventory cert variables are\nregistered to hostvars[octavia_cert_setup_host] i.e hostvars scope.\nCerts facts are set according to octavia_cert_setup_host value.\nCerts distribution play copies certs from certs facts to each octavia\nconatainer.\n\nChange-Id: I0906c7cf80c3345fb993a71fe190fa2f6baa55e4\nCloses-Bug: 1842117\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/13cc76cb8e2a67e5ac2209bd55f3f56b5d87dc73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/13cc76cb8e2a67e5ac2209bd55f3f56b5d87dc73"}]},"branch":"refs/heads/master"},"833b58da2c71f0e6e77253388a64d00d1ccf83f3":{"kind":"REWORK","_number":9,"created":"2019-09-20 13:56:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/62/657662/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/62/657662/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/62/657662/9"}}},"commit":{"parents":[{"commit":"ae5edf5593487bd8f2fb848686ea77ac493eb140","subject":"Merge \"Add cinder volume support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ae5edf5593487bd8f2fb848686ea77ac493eb140"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-09-03 16:05:58.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-20 13:55:48.000000000","tz":180},"subject":"Octavia certificate distribution","message":"Octavia certificate distribution\n\nCurrent copy certificates task works only if octavia_cert_setup_host is\nlocalhost, because it copies from local path. See bug 1842117\n\nTo fix this issue certs are slurped and registered after they are\ncreated.\nIf octavia_cert_setup_host is \u0027localhost\u0027 cert variables are\nregistered to task_vars scope.\nIf octavia_cert_setup_host is one of the octavia containers or utility\ncontainers, i.e. setup host exists in inventory cert variables are\nregistered to hostvars[octavia_cert_setup_host] i.e hostvars scope.\nCerts facts are set according to octavia_cert_setup_host value.\nCerts distribution play copies certs from certs facts to each octavia\nconatainer.\n\nChange-Id: I0906c7cf80c3345fb993a71fe190fa2f6baa55e4\nCloses-Bug: 1842117\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/833b58da2c71f0e6e77253388a64d00d1ccf83f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/833b58da2c71f0e6e77253388a64d00d1ccf83f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
