)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~658034","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~Iaab62fec5571baf73afe365f87635f630128fecf","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Iaab62fec5571baf73afe365f87635f630128fecf","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-09 11:35:18.000000000","updated":"2019-05-09 15:14:40.000000000","submitted":"2019-05-09 15:14:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658034-1557414880902-21cd92bb","meta_rev_id":"38583595cf6cec3df19859fe20310a0b40358e11","_number":658034,"virtual_id_number":658034,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-09 15:14:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-09 12:57:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-09 12:57:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 12:57:13.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-09 15:14:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8a9af696f88f3c887602e78fa7394c8385068329","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-09 11:35:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9ca8411a0f59191f20ce79f7083801548a52145","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-09 12:57:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7d5e7e759c96be70c75a7d8d36c20cdc3cc92a3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:29:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/34/658034/1/check/openstack-ansible-linters/76f3336/ : SUCCESS in 5m 46s\n- openstack-tox-docs http://logs.openstack.org/34/658034/1/check/openstack-tox-docs/4d6d242/html/ : SUCCESS in 2m 15s\n- build-openstack-releasenotes http://logs.openstack.org/34/658034/1/check/build-openstack-releasenotes/f6dea98/html/ : SUCCESS in 2m 12s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/34/658034/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/2d05526/ : SUCCESS in 1h 23m 52s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/34/658034/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/b777795/ : SUCCESS in 1h 04m 20s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/34/658034/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/778b46f/ : SUCCESS in 1h 06m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ddbce4671124795acd4b7858983e89275cfdb987","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:29:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf6e971660774049e5b32ed8f984c45bf8d29419","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 15:14:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fe222da6895a6e72e9bf72b129e5fbb39ff69b47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 15:14:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/34/658034/1/gate/openstack-ansible-linters/d15b86b/ : SUCCESS in 5m 36s\n- openstack-tox-docs http://logs.openstack.org/34/658034/1/gate/openstack-tox-docs/3a07289/html/ : SUCCESS in 2m 53s\n- build-openstack-releasenotes http://logs.openstack.org/34/658034/1/gate/build-openstack-releasenotes/13d67ef/html/ : SUCCESS in 2m 40s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/34/658034/1/gate/openstack-ansible-deploy-aio_metal_octavia-centos-7/8dbb824/ : SUCCESS in 1h 28m 19s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/34/658034/1/gate/openstack-ansible-deploy-aio_metal_octavia-debian-stable/ed954a1/ : SUCCESS in 1h 05m 51s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/34/658034/1/gate/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/d092bdc/ : SUCCESS in 1h 44m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dcdeb98468161402a3585990704749b40cac8ed2","revisions":{"dcdeb98468161402a3585990704749b40cac8ed2":{"kind":"REWORK","_number":1,"created":"2019-05-09 11:35:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/34/658034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/34/658034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/34/658034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/34/658034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/34/658034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/34/658034/1"}}},"commit":{"parents":[{"commit":"9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9481d0d0c8a1f68569a64f1fd1c36a258b00c6e1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:35:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:35:17.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Iaab62fec5571baf73afe365f87635f630128fecf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
