)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~658609","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~stable%2Fstein~I37d1aaa6f3aa97f9344037cdcf9b213f9b5a3f69","project":"openstack/openstack-ansible-os_octavia","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I37d1aaa6f3aa97f9344037cdcf9b213f9b5a3f69","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-11 23:59:24.000000000","updated":"2019-05-12 03:28:57.000000000","submitted":"2019-05-12 03:28:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658609-1557631737527-72f24b91","meta_rev_id":"701bb5cfca8e25f66186becf444176981282a5fe","_number":658609,"virtual_id_number":658609,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-12 03:28:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-12 01:55:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-12 01:55:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-12 01:55:49.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-12 03:28:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07448caff48a25bed272691de24684381f3226f2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-11 23:59:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76e7f4bf77d44ccc0be7011c3b73dae1f811b513","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:23:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/09/658609/1/check/openstack-ansible-linters/d965650/ : SUCCESS in 6m 14s\n- openstack-tox-docs http://logs.openstack.org/09/658609/1/check/openstack-tox-docs/9440a62/html/ : SUCCESS in 2m 36s\n- build-openstack-releasenotes http://logs.openstack.org/09/658609/1/check/build-openstack-releasenotes/bc3493a/html/ : SUCCESS in 3m 23s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/09/658609/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/a579e2a/ : SUCCESS in 1h 18m 20s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/09/658609/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/7e4a860/ : SUCCESS in 59m 06s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/09/658609/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/08a98af/ : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":1},{"id":"9227ffc6afe4ad6b1a10382e622088c19db7a8bd","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-12 01:55:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"37dc5bcef154bd84b64fae1605be8f9ec14c40c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:56:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d7bed91d20f5fc9971cf3734cf8b8483ccd2d22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 03:28:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"159746d790116d45c53eccfc90ebea65f39baf14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 03:28:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/09/658609/1/gate/openstack-ansible-linters/baf04ac/ : SUCCESS in 6m 29s\n- openstack-tox-docs http://logs.openstack.org/09/658609/1/gate/openstack-tox-docs/f37ff71/html/ : SUCCESS in 2m 30s\n- build-openstack-releasenotes http://logs.openstack.org/09/658609/1/gate/build-openstack-releasenotes/854ba6d/html/ : SUCCESS in 2m 03s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/09/658609/1/gate/openstack-ansible-deploy-aio_metal_octavia-centos-7/66968ae/ : SUCCESS in 1h 18m 41s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/09/658609/1/gate/openstack-ansible-deploy-aio_metal_octavia-debian-stable/e214c69/ : SUCCESS in 1h 00m 59s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/09/658609/1/gate/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/6b998a3/ : SUCCESS in 1h 19m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"938f16a745cd783e3a727f49b0bb6be307d00cf7","revisions":{"938f16a745cd783e3a727f49b0bb6be307d00cf7":{"kind":"REWORK","_number":1,"created":"2019-05-11 23:59:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/658609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/09/658609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/09/658609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/09/658609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/09/658609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/09/658609/1"}}},"commit":{"parents":[{"commit":"e8057a584c48dc49494c77bb6ed410b61fca5df8","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e8057a584c48dc49494c77bb6ed410b61fca5df8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:59:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:59:23.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I37d1aaa6f3aa97f9344037cdcf9b213f9b5a3f69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/938f16a745cd783e3a727f49b0bb6be307d00cf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/938f16a745cd783e3a727f49b0bb6be307d00cf7"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
