)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~659754","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I3f277ce4c785b379dbfc76a11759b6d35c1dccea","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"drop_private","hashtags":[],"change_id":"I3f277ce4c785b379dbfc76a11759b6d35c1dccea","subject":"Drop private argument for include/import role","status":"MERGED","created":"2019-05-17 08:55:57.000000000","updated":"2019-06-28 22:21:14.000000000","submitted":"2019-06-28 22:21:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659754-1561760474904-a45a0009","meta_rev_id":"e64247dbca6ce0ab8f537e98f2eefe6eb711dfbb","_number":659754,"virtual_id_number":659754,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"date":"2019-06-17 09:24:25.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-28 22:21:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"recommended":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2019-05-17 11:38:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2019-05-17 23:13:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-28 20:32:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-28 20:32:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 11:38:22.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2019-05-17 23:13:35.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-28 20:32:31.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-28 22:21:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d44fa097d70228d65fc0c2c5d08a7bb8fe9214f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-17 08:55:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5dc3fddd08ec1ebcc80fa13bc376e56da3fde97e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 10:23:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/659754/1/check/openstack-ansible-linters/49aac86/ : SUCCESS in 4m 29s\n- openstack-tox-docs http://logs.openstack.org/54/659754/1/check/openstack-tox-docs/36d1f5d/html/ : SUCCESS in 2m 10s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/f674306/ : SUCCESS in 1h 21m 39s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/71f87fb/ : SUCCESS in 1h 07m 10s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/0a40581/ : FAILURE in 1h 00m 34s","accounts_in_message":[],"_revision_number":1},{"id":"68ee727db346ef684f98b21250c5a61efec25f13","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-05-17 11:38:22.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cdf7adcd9f664d89789ab859fe877990c8a9664a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 13:08:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/659754/1/check/openstack-ansible-linters/31d4d11/ : SUCCESS in 4m 55s\n- openstack-tox-docs http://logs.openstack.org/54/659754/1/check/openstack-tox-docs/22247d3/html/ : SUCCESS in 2m 14s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/c6d6402/ : SUCCESS in 1h 24m 02s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/fc30823/ : FAILURE in 1h 06m 56s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/c8be027/ : FAILURE in 1h 00m 13s","accounts_in_message":[],"_revision_number":1},{"id":"01b05a4bfd51c867a02a755f659700514c6a0163","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-17 14:28:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b0ac8426250bac4c90ffb410f9dbdafd0ce48078","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 16:15:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/659754/1/check/openstack-ansible-linters/d8ec12f/ : SUCCESS in 4m 34s\n- openstack-tox-docs http://logs.openstack.org/54/659754/1/check/openstack-tox-docs/6b2942e/html/ : SUCCESS in 2m 16s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/98bbc63/ : SUCCESS in 1h 23m 39s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/f8d5148/ : SUCCESS in 1h 05m 45s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/903b107/ : FAILURE in 1h 01m 06s","accounts_in_message":[],"_revision_number":1},{"id":"3f0f086f02fcfbb444c50ec1c894824c550d77ea","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-17 23:13:35.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"48c6892f020d4b0d9ee733cda9ef8231e9a9f116","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 00:41:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/659754/1/check/openstack-ansible-linters/13b31c5/ : SUCCESS in 5m 12s\n- openstack-tox-docs http://logs.openstack.org/54/659754/1/check/openstack-tox-docs/044c560/html/ : SUCCESS in 2m 33s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/2bf070c/ : SUCCESS in 1h 22m 28s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/9bc1806/ : SUCCESS in 54m 09s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/1bf932f/ : FAILURE in 1h 00m 09s","accounts_in_message":[],"_revision_number":1},{"id":"0891175e7ace863703bf82c75658366b54f0ded1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-20 08:45:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"96f69544e09bc76a38d432fdc99bb6e5a1732ac8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 10:03:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/659754/1/check/openstack-ansible-linters/6ab496a/ : SUCCESS in 8m 31s\n- openstack-tox-docs http://logs.openstack.org/54/659754/1/check/openstack-tox-docs/9690e73/html/ : SUCCESS in 2m 41s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/6494cb3/ : SUCCESS in 1h 16m 53s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/3fdb826/ : FAILURE in 1h 04m 06s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/68b6e2b/ : FAILURE in 1h 02m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c605321ebfe3476a1063e9540114ae62a21f4b65","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-20 10:15:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dac57132b53ea9ae365409c41c3be20c0031fddc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 11:44:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/659754/1/check/openstack-ansible-linters/7aac2e1/ : SUCCESS in 6m 57s\n- openstack-tox-docs http://logs.openstack.org/54/659754/1/check/openstack-tox-docs/035cd7f/html/ : SUCCESS in 2m 46s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/5216c54/ : SUCCESS in 1h 21m 37s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/d74311a/ : SUCCESS in 1h 11m 33s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/e883e26/ : FAILURE in 1h 00m 18s","accounts_in_message":[],"_revision_number":1},{"id":"fdaf95bbb1340f99d7d9e0b122b48945fd25f0a0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-21 13:20:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"00674c12998f5a1f5f3cd648dfc6d433e5985fd3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 15:10:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/659754/1/check/openstack-ansible-linters/bd86d8a/ : SUCCESS in 6m 06s\n- openstack-tox-docs http://logs.openstack.org/54/659754/1/check/openstack-tox-docs/7e33389/html/ : SUCCESS in 2m 13s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/421c998/ : SUCCESS in 1h 35m 43s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/bbdb8b3/ : SUCCESS in 1h 02m 49s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/d74931c/ : FAILURE in 1h 03m 40s","accounts_in_message":[],"_revision_number":1},{"id":"4c1281d2582eda985261da4c2f06b608757d8870","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-26 12:12:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4a7cb74d1a2f88f308c7d78b388f6affef35d6d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-26 14:21:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/659754/1/check/openstack-ansible-linters/b70d70a/ : SUCCESS in 4m 07s\n- openstack-tox-docs http://logs.openstack.org/54/659754/1/check/openstack-tox-docs/ee0d527/html/ : SUCCESS in 1m 57s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/4001786/ : SUCCESS in 1h 40m 09s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/2374857/ : FAILURE in 1h 02m 26s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/1687636/ : FAILURE in 58m 35s","accounts_in_message":[],"_revision_number":1},{"id":"862e4df3c74d2733f5dcd359d8521c03ea62f5b2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-28 06:58:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6c43160b97afdc3469f5e55be8575205f78a40d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 08:36:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/659754/1/check/openstack-ansible-linters/2935ccf/ : SUCCESS in 4m 27s\n- openstack-tox-docs http://logs.openstack.org/54/659754/1/check/openstack-tox-docs/2abe870/html/ : SUCCESS in 2m 13s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/7e17279/ : SUCCESS in 1h 24m 32s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/9bdf1c5/ : SUCCESS in 1h 02m 22s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/8abf265/ : FAILURE in 1h 21m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a39636eaa3c2450344dcbddcf7ce209b36eab696","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-05 09:54:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e0dc3448f0e5c0dcb205b3a6bec88045b3fe4565","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 11:45:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/659754/1/check/openstack-ansible-linters/74b66d3/ : SUCCESS in 4m 36s\n- openstack-tox-docs http://logs.openstack.org/54/659754/1/check/openstack-tox-docs/5330581/html/ : SUCCESS in 2m 09s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/4f3fa12/ : SUCCESS in 1h 33m 58s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/3d16e01/ : SUCCESS in 1h 08m 11s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/31d8fa3/ : FAILURE in 1h 02m 26s","accounts_in_message":[],"_revision_number":1},{"id":"dec08a59f52d2db562a201c47d9bb491e2b822d0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-10 18:20:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9f24b92b8232bdb2dff07b07c11ad478f737fdac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 19:47:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/659754/1/check/openstack-ansible-linters/4459217/ : SUCCESS in 7m 26s\n- openstack-tox-docs http://logs.openstack.org/54/659754/1/check/openstack-tox-docs/a9019a1/html/ : SUCCESS in 2m 56s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/b857217/ : SUCCESS in 1h 18m 16s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/a3a8ddb/ : SUCCESS in 1h 09m 57s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/ed43d4a/ : FAILURE in 1h 03m 16s","accounts_in_message":[],"_revision_number":1},{"id":"73612b358d95db7cb02c096aa8dc216602451f19","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-17 09:24:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b4df3c16ac7fe09e8e6622bbeb5fe653641367f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 11:18:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/659754/1/check/openstack-ansible-linters/390e416/ : SUCCESS in 4m 44s\n- openstack-tox-docs http://logs.openstack.org/54/659754/1/check/openstack-tox-docs/238e46a/html/ : SUCCESS in 2m 30s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/7416f1f/ : SUCCESS in 1h 53m 26s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/997529c/ : SUCCESS in 1h 06m 19s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/12ace26/ : FAILURE in 1h 04m 40s","accounts_in_message":[],"_revision_number":1},{"id":"94f8bdc4b506ae78b5e148a5be3cf86cb6cfa272","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-28 12:19:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e0be611dc89491bb907aab66627b7c0a21c7d484","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 15:11:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/659754/1/check/openstack-ansible-linters/4b2d281/ : SUCCESS in 5m 51s\n- openstack-tox-docs http://logs.openstack.org/54/659754/1/check/openstack-tox-docs/007dfd0/html/ : SUCCESS in 2m 59s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/8e3c60c/ : SUCCESS in 1h 53m 51s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/ef7a7a7/ : SUCCESS in 1h 21m 10s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/9311b8a/ : SUCCESS in 1h 05m 10s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/7a534bb/ : FAILURE in 49m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e18303b90eba81ecb0d230720f9a7131f5d5bd27","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-28 18:42:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"257f32fd3655ef86fbe7633520cd84156d6cc14b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 20:26:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/54/659754/1/check/openstack-ansible-linters/d9fba2d/ : SUCCESS in 5m 57s\n- openstack-tox-docs http://logs.openstack.org/54/659754/1/check/openstack-tox-docs/047063f/html/ : SUCCESS in 2m 51s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/7317341/ : SUCCESS in 1h 38m 14s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/b393fdd/ : SUCCESS in 1h 07m 40s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/dab2d4d/ : SUCCESS in 1h 01m 13s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/54/659754/1/check/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/4ddf45a/ : SUCCESS in 1h 02m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d4578c22d02ccac5e21492eaa703ae1ffae0710b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-28 20:32:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"95a69010118f9308f078dd3fc35cd8e2323596d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 20:32:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f85a77361252a0398663e07334e01351b120d6df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 22:21:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cebc7fcbb4df9986029f39e227f519f780a7864d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 22:21:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/54/659754/1/gate/openstack-ansible-linters/4672f48/ : SUCCESS in 8m 41s\n- openstack-tox-docs http://logs.openstack.org/54/659754/1/gate/openstack-tox-docs/ce7f324/html/ : SUCCESS in 3m 39s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/54/659754/1/gate/openstack-ansible-deploy-aio_metal_octavia-centos-7/208c04b/ : SUCCESS in 1h 31m 20s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/54/659754/1/gate/openstack-ansible-deploy-aio_metal_octavia-debian-stable/4f36bda/ : SUCCESS in 1h 11m 09s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/54/659754/1/gate/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/db855f4/ : SUCCESS in 1h 23m 28s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/54/659754/1/gate/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/6ebd1f1/ : SUCCESS in 1h 04m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"39343f0576cca7d31c90331084aed0e34bae17fc","revisions":{"39343f0576cca7d31c90331084aed0e34bae17fc":{"kind":"REWORK","_number":1,"created":"2019-05-17 08:55:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/54/659754/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/54/659754/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/54/659754/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/54/659754/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/54/659754/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/54/659754/1"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-05-17 08:55:37.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-05-17 08:55:55.000000000","tz":180},"subject":"Drop private argument for include/import role","message":"Drop private argument for include/import role\n\nSince ansible 2.8 dropped private argument is not supported anymore:\nhttps://github.com/ansible/ansible/issues/45038\n\nChange-Id: I3f277ce4c785b379dbfc76a11759b6d35c1dccea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/39343f0576cca7d31c90331084aed0e34bae17fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/39343f0576cca7d31c90331084aed0e34bae17fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
