)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~660276","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I1a470245626fcea5e9d33cf03787d168cb988f19","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"use-integrated-repo-for-all-tests","hashtags":[],"change_id":"I1a470245626fcea5e9d33cf03787d168cb988f19","subject":"Add integrated coverage","status":"ABANDONED","created":"2019-05-21 08:11:29.000000000","updated":"2019-07-10 15:56:31.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1c8398d93ee91a991a050f25d952ab7efd7903a7","_number":660276,"virtual_id_number":660276,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-06-27 12:33:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-06-28 12:18:53.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 12:33:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-28 12:18:53.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"5878249dd096e7ed5089c49b35fb19cd12fda071","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 08:11:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75d62a97800163f4077cf8cb8436b02eb5eeb6df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 09:28:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/76/660276/1/check/openstack-ansible-linters/9069242/ : SUCCESS in 6m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/76/660276/1/check/openstack-ansible-deploy-aio_metal-debian-stable/45acf67/ : SUCCESS in 1h 04m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/76/660276/1/check/openstack-ansible-deploy-aio_metal-centos-7/3a93f9b/ : SUCCESS in 1h 12m 02s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/76/660276/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/71105cd/ : FAILURE in 48m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/76/660276/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/d538fbc/ : FAILURE in 59m 38s\n- openstack-tox-docs http://logs.openstack.org/76/660276/1/check/openstack-tox-docs/c803eeb/html/ : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1},{"id":"fbbef80a1500ac1f0a6e62eabc223320689de38a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 06:11:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d9a7d51376e06c90b200c9a584560cd0aac52eda","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 08:55:25.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ae7b0d5564508f0401d9f587e1fb8b2f654058ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 12:33:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/76/660276/2/check/openstack-ansible-linters/d6362cb/ : SUCCESS in 5m 42s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/76/660276/2/check/openstack-ansible-deploy-aio_metal-debian-stable/b657593/ : FAILURE in 47m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/76/660276/2/check/openstack-ansible-deploy-aio_metal-centos-7/d236660/ : SUCCESS in 1h 35m 45s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/76/660276/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/a9fb9fd/ : FAILURE in 58m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/76/660276/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/99e073d/ : FAILURE in 58m 11s\n- openstack-tox-docs http://logs.openstack.org/76/660276/2/check/openstack-tox-docs/a20ca97/html/ : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":2},{"id":"96964a72792923524fa0ef46034aa65ec03a6fb2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-28 12:18:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7059a6b8978af0457f0aa4c51ded7553aa355f17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 13:45:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/76/660276/2/check/openstack-ansible-linters/1056c91/ : SUCCESS in 6m 49s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/76/660276/2/check/openstack-ansible-deploy-aio_metal-debian-stable/ac533d3/ : SUCCESS in 1h 12m 46s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/76/660276/2/check/openstack-ansible-deploy-aio_metal-centos-7/661d878/ : SUCCESS in 1h 22m 20s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/76/660276/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/5f95808/ : FAILURE in 57m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/76/660276/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/f9f7a8f/ : SUCCESS in 1h 02m 50s\n- openstack-tox-docs http://logs.openstack.org/76/660276/2/check/openstack-tox-docs/45722db/html/ : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":2},{"id":"64d2014a43f30519b3be0c6d3ab8e3753b4c9d7b","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-10 15:56:31.000000000","message":"Abandoned\n\nSee https://review.opendev.org/#/c/660367/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8a05b6199091b0112c42a046af00940b850830e0","revisions":{"e4389de9db98f1cdbefc6c765f573d5a9e28a130":{"kind":"REWORK","_number":1,"created":"2019-05-21 08:11:29.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/76/660276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/76/660276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/660276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/660276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/660276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/660276/1"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 08:10:52.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 08:11:09.000000000","tz":120},"subject":"Add integrated coverage","message":"Add integrated coverage\n\nCurrently this role doesn\u0027t test the integrated jobs.\nWith the dynamic execution of the roles, this role\nalso get tested by the integrated repo.\n\nChange-Id: I1a470245626fcea5e9d33cf03787d168cb988f19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e4389de9db98f1cdbefc6c765f573d5a9e28a130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e4389de9db98f1cdbefc6c765f573d5a9e28a130"}]},"branch":"refs/heads/master"},"8a05b6199091b0112c42a046af00940b850830e0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-27 08:55:25.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/76/660276/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/76/660276/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/660276/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/660276/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/660276/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/660276/2"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 08:10:52.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-27 08:55:23.000000000","tz":0},"subject":"Add integrated coverage","message":"Add integrated coverage\n\nCurrently this role doesn\u0027t test the integrated jobs.\nWith the dynamic execution of the roles, this role\nalso get tested by the integrated repo.\n\nDepends-On: Ia5fda5d417b79189d048c8891b84d57331df1404\nChange-Id: I1a470245626fcea5e9d33cf03787d168cb988f19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8a05b6199091b0112c42a046af00940b850830e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8a05b6199091b0112c42a046af00940b850830e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
