)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~664337","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~stable%2Frocky~I31307c4fdf319d35e6323796696473384563fcf3","project":"openstack/openstack-ansible-os_octavia","branch":"stable/rocky","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I31307c4fdf319d35e6323796696473384563fcf3","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-10 15:44:57.000000000","updated":"2019-06-12 11:24:39.000000000","submitted":"2019-06-12 11:24:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664337-1560338679381-b5d80b2d","meta_rev_id":"4f1adc63d9d26c07f4a1f6d07ee888bbdfa1594a","_number":664337,"virtual_id_number":664337,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-10 17:39:21.000000000","_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2019-06-12 11:24:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-10 21:11:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-10 21:11:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 17:39:21.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2019-06-10 21:11:00.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-12 11:24:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d6e6028609c701508d7143080f855c1d1e4e9fdf","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:44:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc773689b20118a853b4e4511945f459f8f4401a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:17:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/37/664337/1/check/build-openstack-sphinx-docs/54f0d41/html/ : SUCCESS in 3m 36s\n- openstack-ansible-linters http://logs.openstack.org/37/664337/1/check/openstack-ansible-linters/8f0d74e/ : SUCCESS in 5m 47s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/37/664337/1/check/openstack-ansible-functional-ubuntu-xenial/be7ce2c/ : FAILURE in 1h 06m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b3b86766bfe8dada1fe90e1c56ee4276dcc636d7","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-06-10 17:39:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"27728d4d4c23b306acfd9f4e7c816854401e652b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 18:56:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/37/664337/1/check/build-openstack-sphinx-docs/f016103/html/ : SUCCESS in 4m 27s\n- openstack-ansible-linters http://logs.openstack.org/37/664337/1/check/openstack-ansible-linters/e75a558/ : SUCCESS in 7m 15s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/37/664337/1/check/openstack-ansible-functional-ubuntu-xenial/214f357/ : SUCCESS in 1h 02m 06s","accounts_in_message":[],"_revision_number":1},{"id":"29bb37e9abb4dbc7221c48a9cebd7df93ad8cde3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-10 21:11:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5b281c522bf08bdaa241bdec17cd706d7d2eb0ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 21:11:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc656b48c611b795ae896fa14ffc5bb8ce7a9641","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 22:33:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/37/664337/1/gate/build-openstack-sphinx-docs/443cc1f/html/ : SUCCESS in 3m 33s\n- openstack-ansible-linters http://logs.openstack.org/37/664337/1/gate/openstack-ansible-linters/65ea2b9/ : SUCCESS in 5m 45s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/37/664337/1/gate/openstack-ansible-functional-ubuntu-xenial/8c9b1de/ : FAILURE in 1h 16m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ac96838241fda2a1f36af69930c407650e095a08","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-12 08:39:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"95eb31f92f06daaa159764da8fc1fc26df19a60e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 10:12:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/37/664337/1/check/build-openstack-sphinx-docs/b74dfaf/html/ : SUCCESS in 3m 20s\n- openstack-ansible-linters http://logs.openstack.org/37/664337/1/check/openstack-ansible-linters/6e6d13f/ : SUCCESS in 4m 27s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/37/664337/1/check/openstack-ansible-functional-ubuntu-xenial/6617032/ : SUCCESS in 1h 10m 00s","accounts_in_message":[],"_revision_number":1},{"id":"3c106b615274d6ce51ecb1b1a503e3fa9870f8da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 10:12:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c8a88ad071918d3c93a369b1aeca2f1779dcff3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 11:24:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"340d6ccce809caf27975ad01383ea2cdaa8d4aeb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 11:24:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/37/664337/1/gate/build-openstack-sphinx-docs/d4e27b9/html/ : SUCCESS in 3m 52s\n- openstack-ansible-linters http://logs.openstack.org/37/664337/1/gate/openstack-ansible-linters/794ba54/ : SUCCESS in 4m 38s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/37/664337/1/gate/openstack-ansible-functional-ubuntu-xenial/88ed313/ : SUCCESS in 1h 08m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4cb1eacbaf7d6f3b594459eae70a07a9a004e9d6","revisions":{"4cb1eacbaf7d6f3b594459eae70a07a9a004e9d6":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:44:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/664337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/37/664337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/37/664337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/37/664337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/37/664337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/37/664337/1"}}},"commit":{"parents":[{"commit":"8ca838d1b0b938b46e42a784304771fbca3c78bf","subject":"octavia container network address pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8ca838d1b0b938b46e42a784304771fbca3c78bf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:44:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:44:56.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I31307c4fdf319d35e6323796696473384563fcf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4cb1eacbaf7d6f3b594459eae70a07a9a004e9d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4cb1eacbaf7d6f3b594459eae70a07a9a004e9d6"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
