)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~664551","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I62e4f9d454b75e0fc87a7df2cf70d119e627ea9d","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"fix-octavia","hashtags":[],"change_id":"I62e4f9d454b75e0fc87a7df2cf70d119e627ea9d","subject":"Ensure that admin endpoint is used to configure octavia quota","status":"MERGED","created":"2019-06-11 10:30:45.000000000","updated":"2019-07-01 07:26:47.000000000","submitted":"2019-06-29 09:00:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664551-1561798845198-971b6051","meta_rev_id":"c190f6a73f4e6dcaba7dc472aad711cb66991f8b","_number":664551,"virtual_id_number":664551,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-29 09:00:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-06-28 20:50:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-06-28 21:19:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":1,"date":"2019-06-28 21:19:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-29 07:11:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-28 20:50:35.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-28 21:19:30.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-29 09:00:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"086a979b3870628b6dd986ed93da6a13ba8eb10a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 10:30:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dee0ac9874fbdcee45a61e0112dab39e8c837856","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 13:34:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/51/664551/1/check/openstack-ansible-linters/726ac16/ : SUCCESS in 5m 04s\n- openstack-tox-docs http://logs.openstack.org/51/664551/1/check/openstack-tox-docs/fc05433/html/ : SUCCESS in 2m 29s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/51/664551/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/bcd37e2/ : SUCCESS in 1h 33m 43s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/51/664551/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/6b359e6/ : SUCCESS in 1h 13m 12s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/51/664551/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/fd81dd7/ : TIMED_OUT in 3h 02m 10s","accounts_in_message":[],"_revision_number":1},{"id":"cd19519ac464e98c276b8692f4da6475c2fec22f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 21:20:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d3b79749c926f4d9bf849747e07f22013ec63dea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:51:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/51/664551/1/check/openstack-ansible-linters/95fd06e/ : SUCCESS in 5m 56s\n- openstack-tox-docs http://logs.openstack.org/51/664551/1/check/openstack-tox-docs/66972c5/html/ : SUCCESS in 2m 58s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/51/664551/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/9d24ea1/ : SUCCESS in 1h 25m 32s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/51/664551/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/3f9daa8/ : SUCCESS in 1h 06m 42s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/51/664551/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/e374421/ : FAILURE in 1h 03m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e0660c03ffdfb45af029c30ab95cfeda63533f2c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-26 16:42:24.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"61035b47bba1621f84249bb4ad3f9debf8678edd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 18:23:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/51/664551/2/check/openstack-ansible-linters/d463070/ : SUCCESS in 5m 15s\n- openstack-tox-docs http://logs.openstack.org/51/664551/2/check/openstack-tox-docs/0e19ab2/html/ : SUCCESS in 2m 50s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/51/664551/2/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/3e0d9f0/ : SUCCESS in 1h 33m 51s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/51/664551/2/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/c7e7332/ : SUCCESS in 1h 09m 50s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/51/664551/2/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/a2d2ac2/ : FAILURE in 1h 04m 57s","accounts_in_message":[],"_revision_number":2},{"id":"f8521492010b3ff9e823f00858986936ef769549","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-26 21:12:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6ee5b53750a18147c0310f66cf164658db92dea9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 22:41:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/51/664551/2/check/openstack-ansible-linters/97ce3c8/ : SUCCESS in 6m 17s\n- openstack-tox-docs http://logs.openstack.org/51/664551/2/check/openstack-tox-docs/6687f5a/html/ : SUCCESS in 2m 45s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/51/664551/2/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/39c032a/ : SUCCESS in 1h 23m 10s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/51/664551/2/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/8912dc3/ : SUCCESS in 58m 35s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/51/664551/2/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/7116174/ : FAILURE in 1h 02m 04s","accounts_in_message":[],"_revision_number":2},{"id":"2eb2399a8d7c721d547735287879b73927bce568","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 06:09:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7d2b2305b7f5c851f4992918c8de6fd92fc2873a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 06:16:19.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"e56077a122bcad1ea8f8920502621c0a2ed3596c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 06:25:36.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"bb11cbea1aaffdeada2df216ef2e3d11c4c953b4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 08:53:53.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"f1439cfdc82302c706f12328c6f968de15c564a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 10:38:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/51/664551/4/check/openstack-ansible-linters/8ddcd02/ : SUCCESS in 5m 25s\n- openstack-tox-docs http://logs.openstack.org/51/664551/4/check/openstack-tox-docs/1410c44/html/ : SUCCESS in 2m 36s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/51/664551/4/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/ccbbebf/ : SUCCESS in 1h 31m 31s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/51/664551/4/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/42f0208/ : FAILURE in 59m 02s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/51/664551/4/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/e01f74d/ : FAILURE in 1h 25m 05s","accounts_in_message":[],"_revision_number":4},{"id":"41c8c72074a2ec5e4ffe425b607cd7557f3a0410","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 10:41:49.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"80a5cefac71ed4d8ac4dfaacaaf9ba67da31c621","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:13:36.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/51/664551/5/check/openstack-ansible-linters/fa61b3f/ : SUCCESS in 4m 54s\n- openstack-tox-docs http://logs.openstack.org/51/664551/5/check/openstack-tox-docs/834398f/html/ : SUCCESS in 2m 29s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/51/664551/5/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/e3942f9/ : SUCCESS in 1h 31m 08s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/51/664551/5/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/d09c3ea/ : FAILURE in 1h 10m 52s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/51/664551/5/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/188c7b9/ : FAILURE in 1h 02m 14s","accounts_in_message":[],"_revision_number":5},{"id":"eaa9e3def02c9df615cadb3b8c42e66b7dd2f3fd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 14:12:24.000000000","message":"Topic changed from bug/1831160 to fix-octavia","accounts_in_message":[],"_revision_number":5},{"id":"7d59d249a55e9559087fee8ac9e81e535c488494","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 20:51:52.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"37ec9a88928c13e4038583b977b58ba3978332fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 00:01:18.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/51/664551/6/check/openstack-ansible-linters/2149707/ : SUCCESS in 7m 41s\n- openstack-tox-docs http://logs.openstack.org/51/664551/6/check/openstack-tox-docs/5441f36/html/ : SUCCESS in 3m 07s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/51/664551/6/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/f611586/ : SUCCESS in 1h 24m 51s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/51/664551/6/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/0d73e5f/ : TIMED_OUT in 3h 01m 47s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/51/664551/6/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/e1d93bd/ : FAILURE in 1h 03m 05s","accounts_in_message":[],"_revision_number":6},{"id":"fb8558f69b7186c256c8b2f28414ec87c3868cce","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-28 09:05:18.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"d9ff1d3040bd65cce743a975cfe468ec7dd9859e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 10:43:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/51/664551/7/check/openstack-ansible-linters/11dce55/ : SUCCESS in 5m 05s\n- openstack-tox-docs http://logs.openstack.org/51/664551/7/check/openstack-tox-docs/cd79e8e/html/ : SUCCESS in 2m 53s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/51/664551/7/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/4403350/ : SUCCESS in 1h 31m 03s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/51/664551/7/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/bc4c4ce/ : SUCCESS in 1h 05m 47s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/51/664551/7/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/9cc4133/ : SUCCESS in 1h 01m 42s","accounts_in_message":[],"_revision_number":7},{"id":"402d27f9ba01890dfcf781b663066669c96779b7","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-28 20:50:35.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"de8ae8e0312f9e39e8a16a3659a776b972a01202","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-28 21:19:30.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"a9168346ba75a3dc9ce2dcf01225668bbd378c6e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-29 07:11:49.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"653915ecdd53cc9d1df46043638226d53a5166cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 07:12:00.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"e534b7ad48f4e30ae77b1141146a3803ca0d0c56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 09:00:44.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/51/664551/7/gate/openstack-ansible-linters/ef88e26/ : SUCCESS in 4m 24s\n- openstack-tox-docs http://logs.openstack.org/51/664551/7/gate/openstack-tox-docs/e427257/html/ : SUCCESS in 2m 35s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/51/664551/7/gate/openstack-ansible-deploy-aio_metal_octavia-centos-7/f35cbad/ : SUCCESS in 1h 43m 11s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/51/664551/7/gate/openstack-ansible-deploy-aio_metal_octavia-debian-stable/b0704c5/ : SUCCESS in 1h 06m 38s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/51/664551/7/gate/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/0e9dd15/ : SUCCESS in 1h 00m 29s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/51/664551/7/gate/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/1051f31/ : SUCCESS in 1h 03m 02s","accounts_in_message":[],"_revision_number":7},{"id":"70a42f6bb8b04bbca5d5227aa38e16559d5e4a9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 09:00:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"c190f6a73f4e6dcaba7dc472aad711cb66991f8b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-01 07:26:47.000000000","message":"Patch Set 7: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 54006a86908ecb0c8943165fc3320926826358d7","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"2b39df7d8de3856185ebef0b24382cd0d8b18cca","revisions":{"657dfbda7650131d4fee560114da7b28176d7d52":{"kind":"REWORK","_number":1,"created":"2019-06-11 10:30:45.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/51/664551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/51/664551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/1"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-11 10:28:28.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-11 10:30:41.000000000","tz":60},"subject":"Ensure that admin endpoint is used to configure octavia quota","message":"Ensure that admin endpoint is used to configure octavia quota\n\nThe default is the public endpoint which will fail with an SSL error\nfor AIO or other non-trusted certificates.\n\nChange-Id: I62e4f9d454b75e0fc87a7df2cf70d119e627ea9d\nCloses-Bug: 1831160\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/657dfbda7650131d4fee560114da7b28176d7d52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/657dfbda7650131d4fee560114da7b28176d7d52"}]},"branch":"refs/heads/master"},"1d76fbafea61bf4dde9672649145891836063f72":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-26 16:42:24.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/51/664551/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/51/664551/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/2"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-11 10:28:28.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-26 16:42:21.000000000","tz":0},"subject":"Ensure that admin endpoint is used to configure octavia quota","message":"Ensure that admin endpoint is used to configure octavia quota\n\nThe default is the public endpoint which will fail with an SSL error\nfor AIO or other non-trusted certificates.\n\nDepends-On: I19ef3733c250617a266fce98d11ab29fb548ab16\nChange-Id: I62e4f9d454b75e0fc87a7df2cf70d119e627ea9d\nCloses-Bug: 1831160\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1d76fbafea61bf4dde9672649145891836063f72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1d76fbafea61bf4dde9672649145891836063f72"}]},"branch":"refs/heads/master"},"ed7d43ec9aa97ee8cddd9dc7a23f4e79c2dd3fed":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-06-27 06:25:36.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/51/664551/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/51/664551/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/3"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-11 10:28:28.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-27 06:25:26.000000000","tz":0},"subject":"Ensure that admin endpoint is used to configure octavia quota","message":"Ensure that admin endpoint is used to configure octavia quota\n\nThe default is the public endpoint which will fail with an SSL error\nfor AIO or other non-trusted certificates.\n\nDepends-On: I19ef3733c250617a266fce98d11ab29fb548ab16\nChange-Id: I62e4f9d454b75e0fc87a7df2cf70d119e627ea9d\nCloses-Bug: 1831160\nCloses-Bug: 1828599\nCloses-Bug: 1834394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ed7d43ec9aa97ee8cddd9dc7a23f4e79c2dd3fed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ed7d43ec9aa97ee8cddd9dc7a23f4e79c2dd3fed"}]},"branch":"refs/heads/master"},"412fdf747d4ff2bedfdb32fb24b49d08777524d9":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-06-27 08:53:53.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/51/664551/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/51/664551/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/4"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-11 10:28:28.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-27 08:53:51.000000000","tz":0},"subject":"Ensure that admin endpoint is used to configure octavia quota","message":"Ensure that admin endpoint is used to configure octavia quota\n\nThe default is the public endpoint which will fail with an SSL error\nfor AIO or other non-trusted certificates.\n\nDepends-On: Ia5fda5d417b79189d048c8891b84d57331df1404\nChange-Id: I62e4f9d454b75e0fc87a7df2cf70d119e627ea9d\nCloses-Bug: 1831160\nCloses-Bug: 1828599\nCloses-Bug: 1834394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/412fdf747d4ff2bedfdb32fb24b49d08777524d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/412fdf747d4ff2bedfdb32fb24b49d08777524d9"}]},"branch":"refs/heads/master"},"ee9a170a43e9b5c3ff57fe364ea4d52ba9e50bd6":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-06-27 10:41:49.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/51/664551/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/51/664551/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/5"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-11 10:28:28.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-27 10:41:47.000000000","tz":0},"subject":"Ensure that admin endpoint is used to configure octavia quota","message":"Ensure that admin endpoint is used to configure octavia quota\n\nThe default is the public endpoint which will fail with an SSL error\nfor AIO or other non-trusted certificates.\n\nDepends-On: I3bfe38c74ab5c07db11879706f789d6b701d40e5\nDepends-On: Ia5fda5d417b79189d048c8891b84d57331df1404\nChange-Id: I62e4f9d454b75e0fc87a7df2cf70d119e627ea9d\nCloses-Bug: 1831160\nCloses-Bug: 1828599\nCloses-Bug: 1834394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ee9a170a43e9b5c3ff57fe364ea4d52ba9e50bd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ee9a170a43e9b5c3ff57fe364ea4d52ba9e50bd6"}]},"branch":"refs/heads/master"},"f321355d8e403d9b3d620de62b482ac856ad1a03":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2019-06-27 20:51:52.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/51/664551/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/51/664551/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/6"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-11 10:28:28.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-27 20:51:49.000000000","tz":0},"subject":"Ensure that admin endpoint is used to configure octavia quota","message":"Ensure that admin endpoint is used to configure octavia quota\n\nThe default is the public endpoint which will fail with an SSL error\nfor AIO or other non-trusted certificates.\n\nDepends-On: I3bfe38c74ab5c07db11879706f789d6b701d40e5\nDepends-On: Ia5fda5d417b79189d048c8891b84d57331df1404\nChange-Id: I62e4f9d454b75e0fc87a7df2cf70d119e627ea9d\nCloses-Bug: 1831160\nCloses-Bug: 1828599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f321355d8e403d9b3d620de62b482ac856ad1a03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f321355d8e403d9b3d620de62b482ac856ad1a03"}]},"branch":"refs/heads/master"},"2b39df7d8de3856185ebef0b24382cd0d8b18cca":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2019-06-28 09:05:18.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/51/664551/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/51/664551/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/51/664551/7"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-11 10:28:28.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-28 09:04:39.000000000","tz":0},"subject":"Ensure that admin endpoint is used to configure octavia quota","message":"Ensure that admin endpoint is used to configure octavia quota\n\nThe default is the public endpoint which will fail with an SSL error\nfor AIO or other non-trusted certificates.\n\nDepends-On: I8de48eb1fb4c8d321098ca54b9e21270edc7ac87\nDepends-On: Ide70b5d8f67d8c8a87e3f16671f0f7fb72338b89\nChange-Id: I62e4f9d454b75e0fc87a7df2cf70d119e627ea9d\nCloses-Bug: 1831160\nCloses-Bug: 1828599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/2b39df7d8de3856185ebef0b24382cd0d8b18cca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/2b39df7d8de3856185ebef0b24382cd0d8b18cca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
