)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~664731","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I2d778b43e884139636354d0559ca4462749aa6a3","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I2d778b43e884139636354d0559ca4462749aa6a3","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-11 22:19:01.000000000","updated":"2019-06-28 19:04:45.000000000","submitted":"2019-06-28 19:04:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664731-1561748685670-be057232","meta_rev_id":"15e57cab8b655215ddf2870966911d9a7f4d71a9","_number":664731,"virtual_id_number":664731,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-06-28 19:04:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-28 17:11:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-28 17:11:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 18:54:01.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-28 17:11:29.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-28 19:04:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6b468bf817e908b4ee297d66238a6586106d221","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-11 22:19:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6cad88f6f277686781111ac92b4ec52baab51ee8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 00:21:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/31/664731/1/check/openstack-ansible-linters/66857a0/ : SUCCESS in 4m 29s\n- openstack-tox-docs http://logs.openstack.org/31/664731/1/check/openstack-tox-docs/4276cc9/html/ : SUCCESS in 2m 49s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/31/664731/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/55de25f/ : SUCCESS in 1h 32m 40s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/31/664731/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/b388bc0/ : FAILURE in 1h 02m 54s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/31/664731/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/83ed917/ : FAILURE in 1h 29m 54s","accounts_in_message":[],"_revision_number":1},{"id":"b789090ff1e15715b9a73dc705f5a79ec39ce674","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-18 18:16:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2e37220807cb34622d54deab7a68dc0eead704e1","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-18 18:54:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3d88f2aec371cf726418bcf0edcda803abd9491d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 19:49:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/31/664731/2/check/openstack-ansible-linters/03b6f3f/ : SUCCESS in 4m 45s\n- openstack-tox-docs http://logs.openstack.org/31/664731/2/check/openstack-tox-docs/aadd016/html/ : SUCCESS in 3m 20s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/31/664731/2/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/7496a18/ : SUCCESS in 1h 26m 12s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/31/664731/2/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/824082a/ : SUCCESS in 1h 06m 07s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/31/664731/2/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/6c835de/ : FAILURE in 1h 00m 03s","accounts_in_message":[],"_revision_number":2},{"id":"3a963e5dba1b20323764bd7c7467e7184def4136","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 06:12:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d960146fb238b387bd6b83b2d7512866a5b7c531","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 08:54:29.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"aece9908cb75569730e5d2e7186864bf1d06f437","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-27 10:22:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f3bcece135922a6a8c586650014dc5563d5106fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 12:29:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/31/664731/4/check/openstack-ansible-linters/36be5e7/ : SUCCESS in 8m 01s\n- openstack-tox-docs http://logs.openstack.org/31/664731/4/check/openstack-tox-docs/43f83a5/html/ : SUCCESS in 3m 03s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/31/664731/4/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/e5e3849/ : SUCCESS in 1h 32m 42s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/31/664731/4/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/a0dc559/ : SUCCESS in 1h 10m 35s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/31/664731/4/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/e58d652/ : FAILURE in 1h 04m 31s","accounts_in_message":[],"_revision_number":4},{"id":"4d4b6247cf6ffdc9b7f17397900903b99cf298ef","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-28 12:19:05.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6f7d6b13c6cb91d2048ada997648a11d579001bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 14:03:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/31/664731/4/check/openstack-ansible-linters/e522b80/ : SUCCESS in 5m 39s\n- openstack-tox-docs http://logs.openstack.org/31/664731/4/check/openstack-tox-docs/d5563d1/html/ : SUCCESS in 2m 46s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/31/664731/4/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/5b35864/ : SUCCESS in 1h 35m 30s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/31/664731/4/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/56d3809/ : SUCCESS in 1h 13m 46s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/31/664731/4/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/35b5b6f/ : SUCCESS in 1h 05m 01s","accounts_in_message":[],"_revision_number":4},{"id":"8b355b17aab4e2c47d94472d8c2d35e2469e31bd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-28 17:11:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b5b3379a21c2f1d9f5eb417abc0ac240d28e2e93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 17:11:40.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"95d6281ce2ef31ab4f112f0decd4a85d1455dc2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 19:04:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"87bfbb0c1441b31a3c9d2c309d112da30ecc1998","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 19:04:45.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/31/664731/4/gate/openstack-ansible-linters/4c19bdc/ : SUCCESS in 6m 03s\n- openstack-tox-docs http://logs.openstack.org/31/664731/4/gate/openstack-tox-docs/a12f7b6/html/ : SUCCESS in 2m 58s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/31/664731/4/gate/openstack-ansible-deploy-aio_metal_octavia-centos-7/1b3b822/ : SUCCESS in 1h 52m 48s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/31/664731/4/gate/openstack-ansible-deploy-aio_metal_octavia-debian-stable/1499bd3/ : SUCCESS in 1h 07m 36s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/31/664731/4/gate/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/55bab0b/ : SUCCESS in 1h 12m 45s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/31/664731/4/gate/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/c82c3e9/ : SUCCESS in 1h 11m 53s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9df98e87bb21257adde5a7268fe9db98d67940ad","revisions":{"969f06227353d6c4e3d23f6ebfa89e249b9b20c5":{"kind":"REWORK","_number":1,"created":"2019-06-11 22:19:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/664731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/31/664731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/31/664731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/31/664731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/31/664731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/31/664731/1"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-11 22:18:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-11 22:18:59.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I2d778b43e884139636354d0559ca4462749aa6a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/969f06227353d6c4e3d23f6ebfa89e249b9b20c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/969f06227353d6c4e3d23f6ebfa89e249b9b20c5"}]},"branch":"refs/heads/master"},"114f63a86ba0206f141c3c896f5ea7fc79e7f97b":{"kind":"REWORK","_number":2,"created":"2019-06-18 18:16:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/664731/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/31/664731/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/31/664731/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/31/664731/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/31/664731/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/31/664731/2"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-18 18:16:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-18 18:16:27.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I2d778b43e884139636354d0559ca4462749aa6a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/114f63a86ba0206f141c3c896f5ea7fc79e7f97b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/114f63a86ba0206f141c3c896f5ea7fc79e7f97b"}]},"branch":"refs/heads/master"},"b02d1f6e7ca412b660b516c805d6b1ab97a59d1a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-06-27 08:54:29.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/31/664731/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/31/664731/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/31/664731/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/31/664731/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/31/664731/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/31/664731/3"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-18 18:16:27.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-27 08:54:27.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nDepends-On: Ia5fda5d417b79189d048c8891b84d57331df1404\nChange-Id: I2d778b43e884139636354d0559ca4462749aa6a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b02d1f6e7ca412b660b516c805d6b1ab97a59d1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b02d1f6e7ca412b660b516c805d6b1ab97a59d1a"}]},"branch":"refs/heads/master"},"9df98e87bb21257adde5a7268fe9db98d67940ad":{"kind":"REWORK","_number":4,"created":"2019-06-27 10:22:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/664731/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/31/664731/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/31/664731/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/31/664731/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/31/664731/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/31/664731/4"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-27 10:22:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-27 10:22:48.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I2d778b43e884139636354d0559ca4462749aa6a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9df98e87bb21257adde5a7268fe9db98d67940ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9df98e87bb21257adde5a7268fe9db98d67940ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
