)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~666715","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~stable%2Fstein~If0de68878d0325e6fef9c0a4e700c2fd4c3e9e60","project":"openstack/openstack-ansible-os_octavia","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"If0de68878d0325e6fef9c0a4e700c2fd4c3e9e60","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-20 20:06:14.000000000","updated":"2019-06-24 12:04:09.000000000","submitted":"2019-06-24 12:04:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666715-1561377849509-18733654","meta_rev_id":"bc6d0f1621738ed32226f2f91b51f8b884e68563","_number":666715,"virtual_id_number":666715,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-06-24 12:04:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-06-20 20:09:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-20 22:10:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-20 22:10:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 20:09:11.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-20 22:10:11.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-24 12:04:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"372b70ffec361161b932d19ed340cfb50335e491","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-20 20:06:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5e2329abc30b4ada2f17821234a780b1937ded8","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-20 20:09:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"55819e0abe49ea3b27be5ed16961c94d4b38e82f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 22:02:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/15/666715/1/check/openstack-ansible-linters/e345023/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/15/666715/1/check/openstack-tox-docs/b109d03/html/ : SUCCESS in 3m 30s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/15/666715/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/7fd9943/ : SUCCESS in 1h 18m 38s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/15/666715/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/9bc5fe1/ : SUCCESS in 1h 04m 09s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/15/666715/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/71057ac/ : SUCCESS in 1h 05m 44s","accounts_in_message":[],"_revision_number":1},{"id":"9d314bf78ca415df9520ac0d9ed6aca2c385dc69","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-20 22:10:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"466d083be724048b9f676ed55b4b545fd407f6f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 22:10:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29e4533db923cf9280c48f835b7b6b4385c2c79a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 01:41:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/15/666715/1/gate/openstack-ansible-linters/f733881/ : SUCCESS in 5m 25s\n- openstack-tox-docs http://logs.openstack.org/15/666715/1/gate/openstack-tox-docs/cd5a237/html/ : SUCCESS in 3m 03s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/15/666715/1/gate/openstack-ansible-deploy-aio_metal_octavia-centos-7/eb7fecb/ : SUCCESS in 1h 27m 22s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/15/666715/1/gate/openstack-ansible-deploy-aio_metal_octavia-debian-stable/18150bc/ : TIMED_OUT in 3h 01m 51s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/15/666715/1/gate/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/755bcc0/ : SUCCESS in 1h 00m 13s","accounts_in_message":[],"_revision_number":1},{"id":"bd88278a8523a3c35aa1bb2228c0fc1f1527f205","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-24 09:05:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d32c0006c94716135558297f1b74128522e94c4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 10:37:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/15/666715/1/check/openstack-ansible-linters/0f61411/ : SUCCESS in 5m 52s\n- openstack-tox-docs http://logs.openstack.org/15/666715/1/check/openstack-tox-docs/2ef5add/html/ : SUCCESS in 3m 08s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/15/666715/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/5c3ab26/ : SUCCESS in 1h 23m 52s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/15/666715/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/e387410/ : SUCCESS in 1h 05m 46s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/15/666715/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/726fa82/ : SUCCESS in 1h 00m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d7871b5531faf7f452b29c0c4616cd81987acd5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 10:37:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4291443650661e281d4d240ec9ef854080b8142","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 12:04:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/15/666715/1/gate/openstack-ansible-linters/0095299/ : SUCCESS in 6m 16s\n- openstack-tox-docs http://logs.openstack.org/15/666715/1/gate/openstack-tox-docs/c9f8139/html/ : SUCCESS in 2m 47s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/15/666715/1/gate/openstack-ansible-deploy-aio_metal_octavia-centos-7/f38b061/ : SUCCESS in 1h 20m 18s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/15/666715/1/gate/openstack-ansible-deploy-aio_metal_octavia-debian-stable/d75c5ac/ : SUCCESS in 1h 02m 43s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/15/666715/1/gate/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/8b8bc2b/ : SUCCESS in 59m 39s","accounts_in_message":[],"_revision_number":1},{"id":"62a58a7cd63c7ce5cbeea98342a51259b67cb83d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 12:04:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ace8b076a5cb4b2a9e36a6cc1a86a1f809d5da0e","revisions":{"ace8b076a5cb4b2a9e36a6cc1a86a1f809d5da0e":{"kind":"REWORK","_number":1,"created":"2019-06-20 20:06:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/666715/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/15/666715/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/15/666715/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/15/666715/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/15/666715/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/15/666715/1"}}},"commit":{"parents":[{"commit":"938f16a745cd783e3a727f49b0bb6be307d00cf7","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/938f16a745cd783e3a727f49b0bb6be307d00cf7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-20 20:06:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-20 20:06:12.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: If0de68878d0325e6fef9c0a4e700c2fd4c3e9e60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ace8b076a5cb4b2a9e36a6cc1a86a1f809d5da0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ace8b076a5cb4b2a9e36a6cc1a86a1f809d5da0e"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
