)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~667085","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~Ib945aa3720019893f1f2653484c8374d932daa0d","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"fix-octavia","hashtags":[],"change_id":"Ib945aa3720019893f1f2653484c8374d932daa0d","subject":"Update default test amphora image from xenial to bionic","status":"MERGED","created":"2019-06-24 13:15:03.000000000","updated":"2019-07-01 15:02:05.000000000","submitted":"2019-07-01 15:02:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667085-1561993325540-a45a0009","meta_rev_id":"1802a92cdf3d322f14b766837c1d1fa090899519","_number":667085,"virtual_id_number":667085,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-01 15:02:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2019-07-01 13:19:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2019-06-28 09:20:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":1,"date":"2019-07-01 13:19:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-28 09:20:40.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-07-01 13:19:25.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-07-01 15:02:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"992248a8233b43d0ab4321df097a84e6b1b04508","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-24 13:15:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c78725b4520f5867e6a6c03ea9f7a92c5611a94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 14:46:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/85/667085/1/check/openstack-ansible-linters/069d28f/ : SUCCESS in 5m 04s\n- openstack-tox-docs http://logs.openstack.org/85/667085/1/check/openstack-tox-docs/4d43d44/html/ : SUCCESS in 3m 10s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/85/667085/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/b8f5eda/ : SUCCESS in 1h 27m 45s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/85/667085/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/041d111/ : SUCCESS in 1h 10m 08s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/85/667085/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/f258f54/ : FAILURE in 1h 01m 10s","accounts_in_message":[],"_revision_number":1},{"id":"cd10be1eb5d73456897c08020516fec9cdd23ddd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 06:10:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c2a59722b742c4d2f9947e0f277d2c1950dc16d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 08:00:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/85/667085/1/check/openstack-ansible-linters/13d2326/ : SUCCESS in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/85/667085/1/check/openstack-tox-docs/927bc0e/html/ : SUCCESS in 2m 52s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/85/667085/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/45a94a5/ : SUCCESS in 1h 31m 15s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/85/667085/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/fb47067/ : SUCCESS in 1h 21m 53s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/85/667085/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/ac924d7/ : FAILURE in 1h 02m 09s","accounts_in_message":[],"_revision_number":1},{"id":"2af02507cc57427b1e3dad0d5b71967e7203bf78","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 08:53:15.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2cf14b1fd513608c9af62e641a608f48c28c864c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 10:20:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/85/667085/2/check/openstack-ansible-linters/0534228/ : SUCCESS in 4m 46s\n- openstack-tox-docs http://logs.openstack.org/85/667085/2/check/openstack-tox-docs/d813c79/html/ : SUCCESS in 3m 20s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/85/667085/2/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/a833337/ : SUCCESS in 1h 23m 20s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/85/667085/2/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/128f8e0/ : RETRY_LIMIT in 5m 51s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/85/667085/2/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/5b437f6/ : FAILURE in 1h 02m 36s","accounts_in_message":[],"_revision_number":2},{"id":"c5347b5ef3443c46d07ce0da75b0bafe474bef5f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 14:12:59.000000000","message":"Topic set to fix-octavia","accounts_in_message":[],"_revision_number":2},{"id":"2675aaea61a141125da4b7c170943dee3ec5298b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-28 09:03:58.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"794641db3d864b50d4b4b6f9da79dfea021f0060","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-28 09:20:40.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"33abfaf172c44a9b9dc3e46090a99089376e797b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 10:35:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/85/667085/3/check/openstack-ansible-linters/5963890/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/85/667085/3/check/openstack-tox-docs/6670b70/html/ : SUCCESS in 2m 51s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/85/667085/3/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/5335a65/ : SUCCESS in 1h 28m 47s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/85/667085/3/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/12ac7c8/ : SUCCESS in 1h 08m 37s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/85/667085/3/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/50ed06d/ : SUCCESS in 1h 01m 54s","accounts_in_message":[],"_revision_number":3},{"id":"4f1b69abd341ef83a6749b24c499249d6dc78bd2","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-07-01 13:19:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b2c16edf2c0077569255d9a46a495c6de9a333e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 13:19:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1748da05079a5bad5047dda7e076df84b5edf6f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 15:02:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"1802a92cdf3d322f14b766837c1d1fa090899519","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 15:02:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/85/667085/3/gate/openstack-ansible-linters/7b71a73/ : SUCCESS in 6m 01s\n- openstack-tox-docs http://logs.openstack.org/85/667085/3/gate/openstack-tox-docs/eda4428/html/ : SUCCESS in 3m 22s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/85/667085/3/gate/openstack-ansible-deploy-aio_metal_octavia-centos-7/7d27f22/ : SUCCESS in 1h 19m 42s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/85/667085/3/gate/openstack-ansible-deploy-aio_metal_octavia-debian-stable/3e37690/ : SUCCESS in 1h 06m 51s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/85/667085/3/gate/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/5aca6d4/ : SUCCESS in 1h 07m 18s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/85/667085/3/gate/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/beb8e6f/ : SUCCESS in 1h 12m 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"737e6d165f342b9e8b926cac73cc3ed75dfcfc52","revisions":{"1e4aee7c6682c3666f684898e0049be7eeaff254":{"kind":"REWORK","_number":1,"created":"2019-06-24 13:15:03.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/85/667085/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/85/667085/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/85/667085/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/85/667085/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/85/667085/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/85/667085/1"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-24 13:14:18.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-24 13:14:18.000000000","tz":60},"subject":"Update default test amphora image from xenial to bionic","message":"Update default test amphora image from xenial to bionic\n\nChange-Id: Ib945aa3720019893f1f2653484c8374d932daa0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1e4aee7c6682c3666f684898e0049be7eeaff254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1e4aee7c6682c3666f684898e0049be7eeaff254"}]},"branch":"refs/heads/master"},"052228298bf8f01f72e2efa64564a9dfe1fe2384":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-27 08:53:15.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/85/667085/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/85/667085/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/85/667085/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/85/667085/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/85/667085/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/85/667085/2"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-24 13:14:18.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-27 08:53:01.000000000","tz":0},"subject":"Update default test amphora image from xenial to bionic","message":"Update default test amphora image from xenial to bionic\n\nDepends-On: Ia5fda5d417b79189d048c8891b84d57331df1404\nChange-Id: Ib945aa3720019893f1f2653484c8374d932daa0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/052228298bf8f01f72e2efa64564a9dfe1fe2384"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/052228298bf8f01f72e2efa64564a9dfe1fe2384"}]},"branch":"refs/heads/master"},"737e6d165f342b9e8b926cac73cc3ed75dfcfc52":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-06-28 09:03:58.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/85/667085/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/85/667085/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/85/667085/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/85/667085/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/85/667085/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/85/667085/3"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-24 13:14:18.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-28 09:03:56.000000000","tz":0},"subject":"Update default test amphora image from xenial to bionic","message":"Update default test amphora image from xenial to bionic\n\nDepends-On: I8de48eb1fb4c8d321098ca54b9e21270edc7ac87\nDepends-On: Ide70b5d8f67d8c8a87e3f16671f0f7fb72338b89\nChange-Id: Ib945aa3720019893f1f2653484c8374d932daa0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/737e6d165f342b9e8b926cac73cc3ed75dfcfc52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/737e6d165f342b9e8b926cac73cc3ed75dfcfc52"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
