)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~667791","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~Iddc604e852e0ceb7ef6fd68a123f44a14d8762fa","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"fix-octavia","hashtags":[],"change_id":"Iddc604e852e0ceb7ef6fd68a123f44a14d8762fa","subject":"Use the admin endpoint to upload glance images","status":"MERGED","created":"2019-06-27 06:27:36.000000000","updated":"2019-07-01 07:35:39.000000000","submitted":"2019-06-29 09:00:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667791-1561798846033-1ef4935a","meta_rev_id":"dc82c4522653d22edadffc110549d7eb147395d3","_number":667791,"virtual_id_number":667791,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-29 09:00:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-06-28 20:51:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-06-28 21:16:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":1,"date":"2019-06-28 21:16:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-29 07:12:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-28 20:51:28.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-28 21:16:02.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-29 09:00:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9de4c60a45a03a5ecf06e5cb5fef09b216f868cf","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 06:27:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09b59b2a1af31f2de170d686287092a27429e318","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 06:28:14.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c75658a22d6d5ed8ffbd7fc092b747a75f78c398","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 08:53:30.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"184765110dd5f759f3ae5ea09d0d9614a85fc0e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 10:20:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/91/667791/3/check/openstack-ansible-linters/91777d7/ : SUCCESS in 4m 39s\n- openstack-tox-docs http://logs.openstack.org/91/667791/3/check/openstack-tox-docs/1db4128/html/ : SUCCESS in 2m 31s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/91/667791/3/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/8215204/ : SUCCESS in 1h 22m 27s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/91/667791/3/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/efc19e6/ : FAILURE in 53m 52s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/91/667791/3/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/2f5bd65/ : FAILURE in 1h 01m 25s","accounts_in_message":[],"_revision_number":3},{"id":"03cf2f486b6a5a767100101636cc46cffba3c884","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 14:13:20.000000000","message":"Topic changed from bug/1831183 to fix-octavia","accounts_in_message":[],"_revision_number":3},{"id":"c77ccbac7fc3c0738a70eb87013b93019e3bca60","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 21:07:32.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7942a1b97a382edac005f2b0426530f42cf23d72","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 21:09:02.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"bc85c0024b592e8d55e85446fca41cf521ee5c29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 21:09:19.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"49cbe4e8c783e09b6896d6530d793aaaf2478ce3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-28 07:06:20.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"db7d13ac2a4d28ff958815034addeaed83482506","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 08:40:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/91/667791/4/check/openstack-ansible-linters/7ebbff3/ : SUCCESS in 8m 31s\n- openstack-tox-docs http://logs.openstack.org/91/667791/4/check/openstack-tox-docs/789a3c4/html/ : SUCCESS in 2m 54s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/91/667791/4/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/2509a5a/ : SUCCESS in 1h 28m 29s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/91/667791/4/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/772e1a1/ : SUCCESS in 1h 06m 52s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/91/667791/4/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/9e78f5a/ : SUCCESS in 1h 06m 25s","accounts_in_message":[],"_revision_number":4},{"id":"b8bd5ac081b4a9f2a84ee9f1588173205e50e300","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-28 20:51:28.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6ce15a0387081c33a1f3229c228613372b91ca22","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-28 21:16:02.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"96e99def05e869c54b38f046604d50e1258249f2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-29 07:12:26.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4bae05e7131f1ed32ff39b851ecbf2565c682830","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 07:12:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"545708651bbca8aaf66bbf92f902e9febdf5491e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 09:00:45.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/91/667791/4/gate/openstack-ansible-linters/1c41aa5/ : SUCCESS in 5m 50s\n- openstack-tox-docs http://logs.openstack.org/91/667791/4/gate/openstack-tox-docs/b1a3c38/html/ : SUCCESS in 2m 23s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/91/667791/4/gate/openstack-ansible-deploy-aio_metal_octavia-centos-7/2d0bb16/ : SUCCESS in 1h 23m 05s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/91/667791/4/gate/openstack-ansible-deploy-aio_metal_octavia-debian-stable/293f659/ : SUCCESS in 1h 09m 00s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/91/667791/4/gate/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/4236004/ : SUCCESS in 1h 03m 54s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/91/667791/4/gate/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/99184a1/ : SUCCESS in 1h 06m 30s","accounts_in_message":[],"_revision_number":4},{"id":"07119d6058c0cc7c6336b6c13ef76ed7a0b4a613","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 09:00:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"dc82c4522653d22edadffc110549d7eb147395d3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-01 07:35:39.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit db15fc1728f9d2f73a2ad7be068d6f09187f41eb","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a4d2a2aa64a3076fc5e7a2b1f6b3fea3fd2dd711","revisions":{"926e1b1255f7013d6adb1657e8376f094299abe6":{"kind":"REWORK","_number":1,"created":"2019-06-27 06:27:36.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/91/667791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/91/667791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/91/667791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/91/667791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/91/667791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/91/667791/1"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-27 06:22:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-27 06:22:21.000000000","tz":60},"subject":"Use the admin endpoint to upload glance images.","message":"Use the admin endpoint to upload glance images.\n\nThe deploy host may not have access to the public endpoint. The\npublic endpoint SSL certificate may not be trusted. Both of\nthese things will cause a failure to upload the image.\n\nChange-Id: Iddc604e852e0ceb7ef6fd68a123f44a14d8762fa\nCloses-Bug: 1831183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/926e1b1255f7013d6adb1657e8376f094299abe6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/926e1b1255f7013d6adb1657e8376f094299abe6"}]},"branch":"refs/heads/master"},"7970fefafb5c7b723bc332103dfd926b188a9e82":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-27 06:28:14.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/91/667791/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/91/667791/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/91/667791/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/91/667791/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/91/667791/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/91/667791/2"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-27 06:22:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-27 06:28:12.000000000","tz":0},"subject":"Use the admin endpoint to upload glance images","message":"Use the admin endpoint to upload glance images\n\nThe deploy host may not have access to the public endpoint. The\npublic endpoint SSL certificate may not be trusted. Both of\nthese things will cause a failure to upload the image.\n\nChange-Id: Iddc604e852e0ceb7ef6fd68a123f44a14d8762fa\nCloses-Bug: 1831183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7970fefafb5c7b723bc332103dfd926b188a9e82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7970fefafb5c7b723bc332103dfd926b188a9e82"}]},"branch":"refs/heads/master"},"53cf8cfec9cf48486848640808aab1730c5a79a6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-06-27 08:53:30.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/91/667791/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/91/667791/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/91/667791/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/91/667791/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/91/667791/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/91/667791/3"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-27 06:22:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-27 08:53:28.000000000","tz":0},"subject":"Use the admin endpoint to upload glance images","message":"Use the admin endpoint to upload glance images\n\nThe deploy host may not have access to the public endpoint. The\npublic endpoint SSL certificate may not be trusted. Both of\nthese things will cause a failure to upload the image.\n\nDepends-On: Ia5fda5d417b79189d048c8891b84d57331df1404\nChange-Id: Iddc604e852e0ceb7ef6fd68a123f44a14d8762fa\nCloses-Bug: 1831183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/53cf8cfec9cf48486848640808aab1730c5a79a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/53cf8cfec9cf48486848640808aab1730c5a79a6"}]},"branch":"refs/heads/master"},"a4d2a2aa64a3076fc5e7a2b1f6b3fea3fd2dd711":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-06-27 21:09:02.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/91/667791/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/91/667791/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/91/667791/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/91/667791/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/91/667791/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/91/667791/4"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-27 06:22:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-27 21:09:00.000000000","tz":0},"subject":"Use the admin endpoint to upload glance images","message":"Use the admin endpoint to upload glance images\n\nThe deploy host may not have access to the public endpoint. The\npublic endpoint SSL certificate may not be trusted. Both of\nthese things will cause a failure to upload the image.\n\nDepends-On: I8de48eb1fb4c8d321098ca54b9e21270edc7ac87\nDepends-On: Ide70b5d8f67d8c8a87e3f16671f0f7fb72338b89\nDepends-On: Ia5fda5d417b79189d048c8891b84d57331df1404\nChange-Id: Iddc604e852e0ceb7ef6fd68a123f44a14d8762fa\nCloses-Bug: 1831183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a4d2a2aa64a3076fc5e7a2b1f6b3fea3fd2dd711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a4d2a2aa64a3076fc5e7a2b1f6b3fea3fd2dd711"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
