)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~668137","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~Id066a2fa2107e9eee741f011f869390b90e05664","project":"openstack/openstack-ansible-os_octavia","branch":"master","hashtags":[],"change_id":"Id066a2fa2107e9eee741f011f869390b90e05664","subject":"[DNM] test patch","status":"ABANDONED","created":"2019-06-28 10:08:05.000000000","updated":"2019-07-10 10:40:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4c4a688ee42953652462b2d4e2b01b87044f78cc","_number":668137,"virtual_id_number":668137,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-06-28 13:16:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-28 13:16:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f5d1f2e15911c96edc6db8f3af3cf8c54b0c693d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-28 10:08:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7934911227a18599880174ac31345aaab97ab8e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 13:16:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/37/668137/1/check/openstack-ansible-linters/1bf9f05/ : SUCCESS in 6m 07s\n- openstack-tox-docs http://logs.openstack.org/37/668137/1/check/openstack-tox-docs/19d652b/html/ : SUCCESS in 2m 49s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/37/668137/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/9e1be50/ : SUCCESS in 1h 33m 48s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/37/668137/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/39f65ca/ : FAILURE in 56m 53s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/37/668137/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/455e8ca/ : TIMED_OUT in 3h 03m 08s","accounts_in_message":[],"_revision_number":1},{"id":"4c4a688ee42953652462b2d4e2b01b87044f78cc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-10 10:40:54.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"502efea17047aa75cbae3ec89f380a4b2a6040c5","revisions":{"502efea17047aa75cbae3ec89f380a4b2a6040c5":{"kind":"REWORK","_number":1,"created":"2019-06-28 10:08:05.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/37/668137/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/37/668137/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/37/668137/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/37/668137/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/37/668137/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/37/668137/1"}}},"commit":{"parents":[{"commit":"dcdeb98468161402a3585990704749b40cac8ed2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dcdeb98468161402a3585990704749b40cac8ed2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-28 10:07:14.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-28 10:07:50.000000000","tz":60},"subject":"[DNM] test patch","message":"[DNM] test patch\n\nDepends-On: I89232f7bfac2398796550880396552a139d14fbc\nChange-Id: Id066a2fa2107e9eee741f011f869390b90e05664\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/502efea17047aa75cbae3ec89f380a4b2a6040c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/502efea17047aa75cbae3ec89f380a4b2a6040c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
