)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~668369","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~stable%2Fstein~I62e4f9d454b75e0fc87a7df2cf70d119e627ea9d","project":"openstack/openstack-ansible-os_octavia","branch":"stable/stein","topic":"fix-octavia-stable/stein","hashtags":[],"change_id":"I62e4f9d454b75e0fc87a7df2cf70d119e627ea9d","subject":"Ensure that admin endpoint is used to configure octavia quota","status":"MERGED","created":"2019-07-01 07:26:47.000000000","updated":"2019-07-23 18:04:33.000000000","submitted":"2019-07-23 18:04:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668369-1563905073999-b5d406da","meta_rev_id":"750a64ceca1aa9f5af697887cd91fd1689fa9742","_number":668369,"virtual_id_number":668369,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-23 18:04:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"value":0,"date":"2019-07-23 15:09:47.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"all":[{"value":1,"date":"2019-07-13 18:53:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-07-23 07:51:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-12 22:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-15 22:39:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":1,"date":"2019-07-10 13:22:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":1,"date":"2019-07-13 09:05:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2019-07-23 07:51:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 13:22:03.000000000","updated_by":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"reviewer":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"state":"REVIEWER"},{"updated":"2019-07-12 22:57:38.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-13 09:05:24.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-07-13 18:53:02.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-07-15 22:39:36.000000000","updated_by":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"reviewer":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"state":"REVIEWER"},{"updated":"2019-07-23 07:51:17.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-07-23 18:04:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6f8cde9f8d49dfa83f900eb72d2949777b6e7179","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-01 07:26:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"978daf403aac48d62f75b74cfae816c128090097","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 08:54:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/69/668369/1/check/openstack-ansible-linters/d6e3da2/ : SUCCESS in 6m 30s\n- openstack-tox-docs http://logs.openstack.org/69/668369/1/check/openstack-tox-docs/c2b5e16/html/ : SUCCESS in 3m 05s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/69/668369/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/78c70a3/ : SUCCESS in 1h 20m 56s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/69/668369/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/c0ea3cb/ : SUCCESS in 1h 03m 40s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/69/668369/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/709e143/ : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":1},{"id":"e570a98e80f13e0b0f8d19ca7b06b6b8eea8e435","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-07-10 13:22:03.000000000","message":"Patch Set 1: Code-Review+1\n\nneed this","accounts_in_message":[],"_revision_number":1},{"id":"dbdcfddebb2c4989b52ccd702deb4979ba42ef55","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-12 22:57:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bf7105b059f3a18d0c15591e3f3079b49b8b031d","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-07-13 09:05:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"560530fb26574d60e6e5063e890a30ab8f04f8f6","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-07-13 18:53:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bb994496e99e6b0b9cbc3cc10bc44073bd59aeec","author":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"date":"2019-07-15 22:39:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3a076cd5a228fc5c2b833976c5ba5278b0341ff0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-23 07:51:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8fb100a355cd7e471c10afbcf671ae6306cd475d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 07:51:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d43efa8d10ddd0dfeec7500b948f0149f57c1fe0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 09:30:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/69/668369/1/gate/openstack-ansible-linters/6528f1e/ : SUCCESS in 7m 09s\n- openstack-tox-docs http://logs.openstack.org/69/668369/1/gate/openstack-tox-docs/074cc34/html/ : SUCCESS in 3m 26s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/69/668369/1/gate/openstack-ansible-deploy-aio_metal_octavia-centos-7/821d206/ : FAILURE in 4m 25s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/69/668369/1/gate/openstack-ansible-deploy-aio_metal_octavia-debian-stable/d7a9dc5/ : SUCCESS in 1h 12m 39s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/69/668369/1/gate/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/75c513e/ : SUCCESS in 1h 07m 48s","accounts_in_message":[],"_revision_number":1},{"id":"707e65f95f48c0ccd795d73eb1068e6203fde05d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-23 10:30:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aa17ca48b26976e371c55ba6622781f2f332603a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 11:56:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/69/668369/1/check/openstack-ansible-linters/642afef/ : SUCCESS in 5m 11s\n- openstack-tox-docs http://logs.openstack.org/69/668369/1/check/openstack-tox-docs/de917ed/html/ : SUCCESS in 4m 02s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/69/668369/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/0d32fb1/ : SUCCESS in 1h 21m 04s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/69/668369/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/152f411/ : SUCCESS in 1h 08m 43s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/69/668369/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/8a87749/ : SUCCESS in 58m 55s","accounts_in_message":[],"_revision_number":1},{"id":"066d377b10ddc565c34f633d79b27c070e26e1a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 11:56:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f8f84c9526548688f108d1ebc2e79b694579eb0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 13:13:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/69/668369/1/gate/openstack-ansible-linters/359dffd/ : SUCCESS in 6m 16s\n- openstack-tox-docs http://logs.openstack.org/69/668369/1/gate/openstack-tox-docs/cd6b609/html/ : SUCCESS in 3m 37s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/69/668369/1/gate/openstack-ansible-deploy-aio_metal_octavia-centos-7/8166cfd/ : SUCCESS in 1h 13m 04s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/69/668369/1/gate/openstack-ansible-deploy-aio_metal_octavia-debian-stable/b8cf8d9/ : SUCCESS in 53m 16s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/69/668369/1/gate/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/c952678/ : FAILURE in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"515d6af043b90b8596beae712b594466a0bb9da0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-23 15:09:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"73aa17211d27dd921156d4eece592d838bb63f95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 16:39:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/69/668369/1/check/openstack-ansible-linters/1fa980e/ : SUCCESS in 7m 23s\n- openstack-tox-docs http://logs.openstack.org/69/668369/1/check/openstack-tox-docs/6400a16/html/ : SUCCESS in 2m 40s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/69/668369/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/95d84de/ : SUCCESS in 1h 21m 46s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/69/668369/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/d8e5b34/ : SUCCESS in 1h 17m 40s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/69/668369/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/b869fe5/ : SUCCESS in 1h 00m 34s","accounts_in_message":[],"_revision_number":1},{"id":"faffd2dbbbadd1eff7ca88defbf9e849e718008f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 16:39:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf597d23fae93f74a65277aa4e0be3eb8161f339","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 18:04:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/69/668369/1/gate/openstack-ansible-linters/5107416/ : SUCCESS in 6m 38s\n- openstack-tox-docs http://logs.openstack.org/69/668369/1/gate/openstack-tox-docs/7ada85d/html/ : SUCCESS in 3m 06s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/69/668369/1/gate/openstack-ansible-deploy-aio_metal_octavia-centos-7/fdfefb2/ : SUCCESS in 1h 18m 13s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/69/668369/1/gate/openstack-ansible-deploy-aio_metal_octavia-debian-stable/f067dd6/ : SUCCESS in 1h 20m 01s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/69/668369/1/gate/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/de7f4cf/ : SUCCESS in 1h 19m 47s","accounts_in_message":[],"_revision_number":1},{"id":"dfa1f4068de4c2bdf8a73f290e9836ef640e0ac1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 18:04:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"54006a86908ecb0c8943165fc3320926826358d7","revisions":{"54006a86908ecb0c8943165fc3320926826358d7":{"kind":"REWORK","_number":1,"created":"2019-07-01 07:26:47.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/69/668369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/69/668369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/69/668369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/69/668369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/69/668369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/69/668369/1"}}},"commit":{"parents":[{"commit":"ace8b076a5cb4b2a9e36a6cc1a86a1f809d5da0e","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ace8b076a5cb4b2a9e36a6cc1a86a1f809d5da0e"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-11 10:28:28.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-01 07:26:47.000000000","tz":0},"subject":"Ensure that admin endpoint is used to configure octavia quota","message":"Ensure that admin endpoint is used to configure octavia quota\n\nThe default is the public endpoint which will fail with an SSL error\nfor AIO or other non-trusted certificates.\n\nDepends-On: I8de48eb1fb4c8d321098ca54b9e21270edc7ac87\nDepends-On: Ide70b5d8f67d8c8a87e3f16671f0f7fb72338b89\nChange-Id: I62e4f9d454b75e0fc87a7df2cf70d119e627ea9d\nCloses-Bug: 1831160\nCloses-Bug: 1828599\n(cherry picked from commit 2b39df7d8de3856185ebef0b24382cd0d8b18cca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/54006a86908ecb0c8943165fc3320926826358d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/54006a86908ecb0c8943165fc3320926826358d7"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
