)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~671019","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I8cfdd46a57a91ef3b1879bf85b03cced74234451","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"use_journalctl","hashtags":[],"change_id":"I8cfdd46a57a91ef3b1879bf85b03cced74234451","subject":"Use systemd-journald instead of log files","status":"MERGED","created":"2019-07-16 12:20:47.000000000","updated":"2019-07-17 06:57:38.000000000","submitted":"2019-07-17 06:57:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671019-1563346658315-77280bbb","meta_rev_id":"dac5c6b249c339d9fb535ee300bd8af38f3b250f","_number":671019,"virtual_id_number":671019,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-17 06:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-16 21:34:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-17 05:18:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-17 05:18:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 21:34:03.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-17 05:18:18.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-17 06:57:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"69f7cf294b9c2c550c7982b6ffd5aa04ebac456a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-16 12:20:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"403b56a6eb99d6aa47f83b8453ea3f26553c9b1e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-16 12:58:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"06c5b922277b97a11acbdf4b7b76660b97de51ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 14:32:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/19/671019/2/check/openstack-ansible-linters/307a1f6/ : SUCCESS in 4m 58s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/19/671019/2/check/openstack-ansible-deploy-aio_metal-debian-stable/3039ac3/ : SUCCESS in 1h 12m 13s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/19/671019/2/check/openstack-ansible-deploy-aio_metal-centos-7/ce347a4/ : SUCCESS in 1h 28m 18s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/19/671019/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/6d1fcfb/ : SUCCESS in 1h 14m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/19/671019/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/f9df7ea/ : SUCCESS in 1h 17m 54s\n- openstack-tox-docs http://logs.openstack.org/19/671019/2/check/openstack-tox-docs/c758d65/html/ : SUCCESS in 3m 50s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/19/671019/2/check/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/19346a0/ : SUCCESS in 1h 05m 42s","accounts_in_message":[],"_revision_number":2},{"id":"e38127a0433d9b07bc0034bb87a72becbf36af65","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-16 19:26:24.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0a8aec13687f142fafb1fb1f5362564618afc284","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-16 21:19:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4b316532da38667b5347914103a683dd4a17b64e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-16 21:19:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"551c55576754681128e6aa3c4508b8a543054fc4","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-16 21:34:03.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fe0bc542a45df4c5acff593a78020a0a771e38d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 23:01:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/19/671019/3/check/openstack-ansible-linters/fe83199/ : SUCCESS in 5m 51s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/19/671019/3/check/openstack-ansible-deploy-aio_metal-debian-stable/c340b58/ : SUCCESS in 1h 03m 42s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/19/671019/3/check/openstack-ansible-deploy-aio_metal-centos-7/734f16b/ : SUCCESS in 1h 17m 43s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/19/671019/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/f8adc5a/ : SUCCESS in 1h 34m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/19/671019/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/361cc02/ : SUCCESS in 1h 17m 02s\n- openstack-tox-docs http://logs.openstack.org/19/671019/3/check/openstack-tox-docs/c69ffbd/html/ : SUCCESS in 2m 45s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/19/671019/3/check/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/8e3b533/ : SUCCESS in 1h 08m 30s","accounts_in_message":[],"_revision_number":3},{"id":"6717f0464883ff142427a4f13c238a91c5721111","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-17 05:18:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e226ce718ff44f8add18ac458075137514fcfeb1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 05:18:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"069d0d6f5ffbee0dad823bab0602e329b1e0820a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 06:57:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/19/671019/3/gate/openstack-ansible-linters/8283b81/ : SUCCESS in 4m 51s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/19/671019/3/gate/openstack-ansible-deploy-aio_metal-debian-stable/e49d86b/ : SUCCESS in 1h 11m 13s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/19/671019/3/gate/openstack-ansible-deploy-aio_metal-centos-7/c020ed6/ : SUCCESS in 1h 33m 51s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/19/671019/3/gate/openstack-ansible-deploy-aio_metal-opensuse-15/074e896/ : SUCCESS in 1h 08m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/19/671019/3/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/76b68a8/ : SUCCESS in 57m 54s\n- openstack-tox-docs http://logs.openstack.org/19/671019/3/gate/openstack-tox-docs/e4c1494/html/ : SUCCESS in 2m 42s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 http://logs.openstack.org/19/671019/3/gate/openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7/e10c050/ : SUCCESS in 1h 05m 00s","accounts_in_message":[],"_revision_number":3},{"id":"15e24ce4cba34784897328c47f464f8ec8072207","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 06:57:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"048a43bd81e7b945e4224de722ba2d0239161f1e","revisions":{"4cf344f45aa7ab9a8153fe29420b13a90d9b3104":{"kind":"REWORK","_number":1,"created":"2019-07-16 12:20:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/671019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/671019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/671019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/671019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/671019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/671019/1"}}},"commit":{"parents":[{"commit":"9d86b5f0aae7d679db48a694a9a2d596b873c537","subject":"Merge \"Cap sphinx for py2 to match global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9d86b5f0aae7d679db48a694a9a2d596b873c537"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-16 12:20:21.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-16 12:20:44.000000000","tz":180},"subject":"Use systemd-journald instead of log files","message":"Use systemd-journald instead of log files\n\nThis patch aims to migrate service from usage of regular syslog files\nto journald.\n\nDepends-On: https://review.opendev.org/670824/\nChange-Id: I8cfdd46a57a91ef3b1879bf85b03cced74234451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4cf344f45aa7ab9a8153fe29420b13a90d9b3104"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4cf344f45aa7ab9a8153fe29420b13a90d9b3104"}]},"branch":"refs/heads/master"},"2460b70cf2325880534c626f7fd14242ce33e510":{"kind":"REWORK","_number":2,"created":"2019-07-16 12:58:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/671019/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/671019/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/671019/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/671019/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/671019/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/671019/2"}}},"commit":{"parents":[{"commit":"9d86b5f0aae7d679db48a694a9a2d596b873c537","subject":"Merge \"Cap sphinx for py2 to match global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9d86b5f0aae7d679db48a694a9a2d596b873c537"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-16 12:20:21.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-16 12:57:06.000000000","tz":180},"subject":"Use systemd-journald instead of log files","message":"Use systemd-journald instead of log files\n\nThis patch aims to migrate service from usage of regular syslog files\nto journald. We also disable uwsgi logging, since it dublicates\nrequests that are logged by service itself.\n\nDepends-On: https://review.opendev.org/670824/\nChange-Id: I8cfdd46a57a91ef3b1879bf85b03cced74234451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/2460b70cf2325880534c626f7fd14242ce33e510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/2460b70cf2325880534c626f7fd14242ce33e510"}]},"branch":"refs/heads/master"},"048a43bd81e7b945e4224de722ba2d0239161f1e":{"kind":"REWORK","_number":3,"created":"2019-07-16 21:19:09.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/671019/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/19/671019/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/671019/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/671019/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/671019/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/19/671019/3"}}},"commit":{"parents":[{"commit":"9d86b5f0aae7d679db48a694a9a2d596b873c537","subject":"Merge \"Cap sphinx for py2 to match global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9d86b5f0aae7d679db48a694a9a2d596b873c537"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-16 12:20:21.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-16 21:18:58.000000000","tz":180},"subject":"Use systemd-journald instead of log files","message":"Use systemd-journald instead of log files\n\nThis patch aims to migrate service from usage of regular syslog files\nto journald. We also disable uwsgi logging, since it dublicates\nrequests that are logged by service itself.\n\nDepends-On: https://review.opendev.org/670824/\nChange-Id: I8cfdd46a57a91ef3b1879bf85b03cced74234451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/048a43bd81e7b945e4224de722ba2d0239161f1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/048a43bd81e7b945e4224de722ba2d0239161f1e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
