)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~672825","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~stable%2Fstein~I26f838490d1f6ed08028d9c70d5202fe27ec0a74","project":"openstack/openstack-ansible-os_octavia","branch":"stable/stein","hashtags":[],"change_id":"I26f838490d1f6ed08028d9c70d5202fe27ec0a74","subject":"Fixed octavia-health-manager group name","status":"MERGED","created":"2019-07-25 21:03:56.000000000","updated":"2019-07-30 20:25:22.000000000","submitted":"2019-07-30 20:25:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672825-1564518322023-fe339b59","meta_rev_id":"2b185aad01416b9a8d002a1808b3a3f52cc6f5c2","_number":672825,"virtual_id_number":672825,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-30 20:25:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"value":2,"date":"2019-07-30 12:43:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-30 17:13:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"date":"2019-07-25 21:03:56.000000000","_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2019-07-30 17:13:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-25 21:03:56.000000000","updated_by":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"reviewer":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"state":"REVIEWER"},{"updated":"2019-07-30 12:43:28.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-07-30 17:13:47.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-30 20:25:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e301523ff613394598213c62e8f9e2216b7b0e46","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-25 21:03:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"83e47ed1dfbd0307699ca57f7c0170c76708d24c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-25 21:04:02.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit c9d9b6a69402991a142e8b1c1585a62c3957a616","accounts_in_message":[],"_revision_number":1},{"id":"bdf824941cf81b75269fe0701aee706f12c812ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 22:53:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/25/672825/1/check/openstack-ansible-linters/8c753cb/ : SUCCESS in 7m 23s\n- openstack-tox-docs http://logs.openstack.org/25/672825/1/check/openstack-tox-docs/51e2ca5/html/ : SUCCESS in 3m 04s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/25/672825/1/check/openstack-ansible-deploy-aio_metal_octavia-centos-7/d7a6308/ : SUCCESS in 1h 47m 23s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/25/672825/1/check/openstack-ansible-deploy-aio_metal_octavia-debian-stable/10f71a0/ : SUCCESS in 1h 03m 07s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/25/672825/1/check/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/84b04a7/ : SUCCESS in 1h 10m 26s","accounts_in_message":[],"_revision_number":1},{"id":"b857877359616bb7c407510285c880094b148dc9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-30 12:43:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"57dcb3cb4986bd67c25cacf64a6897bb5d622ad7","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-30 17:13:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"56fd1374863ee21742f0eae37970ab2404c70328","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 17:13:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c2f3defeb1d25f473de7ce22c870e9c1bd8730c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:25:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/25/672825/1/gate/openstack-ansible-linters/80da9a8/ : SUCCESS in 5m 32s\n- openstack-tox-docs http://logs.openstack.org/25/672825/1/gate/openstack-tox-docs/4518ac2/html/ : SUCCESS in 3m 36s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 http://logs.openstack.org/25/672825/1/gate/openstack-ansible-deploy-aio_metal_octavia-centos-7/44ac4e1/ : SUCCESS in 1h 42m 17s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable http://logs.openstack.org/25/672825/1/gate/openstack-ansible-deploy-aio_metal_octavia-debian-stable/33d2084/ : SUCCESS in 58m 43s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic http://logs.openstack.org/25/672825/1/gate/openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic/cc60068/ : SUCCESS in 1h 12m 41s","accounts_in_message":[],"_revision_number":1},{"id":"2b185aad01416b9a8d002a1808b3a3f52cc6f5c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:25:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0cfcadf0a81911b848cf69171505a54ce778e920","revisions":{"0cfcadf0a81911b848cf69171505a54ce778e920":{"kind":"REWORK","_number":1,"created":"2019-07-25 21:03:56.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/25/672825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/25/672825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/25/672825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/25/672825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/25/672825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/25/672825/1"}}},"commit":{"parents":[{"commit":"dd51b6e6bb33c8910d9ad8f33c169b18956453da","subject":"Merge \"Ensure that admin endpoint is used to configure octavia quota\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/dd51b6e6bb33c8910d9ad8f33c169b18956453da"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-07-23 19:36:34.000000000","tz":-240},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-25 21:03:56.000000000","tz":0},"subject":"Fixed octavia-health-manager group name","message":"Fixed octavia-health-manager group name\n\nChange-Id: I26f838490d1f6ed08028d9c70d5202fe27ec0a74\n(cherry picked from commit f28f02e96ccf5d0a57607ab4c04c75c2cd9bd964)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/0cfcadf0a81911b848cf69171505a54ce778e920"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/0cfcadf0a81911b848cf69171505a54ce778e920"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
