)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~679304","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I011662ad85bc3ac2a9f405e5e0df0af1e050544f","project":"openstack/openstack-ansible-os_octavia","branch":"master","hashtags":[],"change_id":"I011662ad85bc3ac2a9f405e5e0df0af1e050544f","subject":"Enable amphora nova flavor extra_specs","status":"MERGED","created":"2019-08-29 16:37:29.000000000","updated":"2019-09-06 11:46:28.000000000","submitted":"2019-09-06 01:42:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679304-1567734150908-7897dbc5","meta_rev_id":"e6f2c40d2769c6389435b08407762cd431cb1116","_number":679304,"virtual_id_number":679304,"owner":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-04 12:11:43.000000000","_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-09-06 01:42:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"value":2,"date":"2019-09-03 17:05:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-30 21:11:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"value":1,"date":"2019-09-03 17:05:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-30 21:11:16.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-09-03 17:05:24.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-09-06 01:42:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee09f5f2e438e1dae7474c7aab5d771661e50cb3","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-08-29 16:37:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2bda9e1db05946307d29ee785a977a95b99b0ad","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-08-29 17:16:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d04daa20465332d668792b350a7a508c08c37375","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 18:42:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/336f504bd0fe46a8afe1afcc454eb827 : SUCCESS in 6m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/09d5b0de20a343cc83007fda008071ac : SUCCESS in 1h 25m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/219e60fd0db941d39a484589c05899a6 : SUCCESS in 2h 01m 44s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/aba629b1eef0454f8a85b4ad0003fa47 : SUCCESS in 1h 24m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88b834da841643f5bc6814c05ad6c55f : SUCCESS in 1h 34m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9fb25af0bf241329cbd83d72047aa29 : SUCCESS in 4m 46s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/ee74c0960f534e328192538d6fe9df99 : SUCCESS in 1h 06m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d09f33984439c9f908ca9d8e06b97d1877861778","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-08-29 21:28:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"784dd8b87770420326134d63eda2ce0e8d01a40d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 23:19:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3df978848c2b4dc9b18e1d9009db087c : SUCCESS in 7m 29s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/f77995f86afd4f62bacb51347e90caa5 : SUCCESS in 1h 17m 03s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1c53cfc25ed14fd18d2e891fa7b34e27 : SUCCESS in 1h 42m 36s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/302415743d554febbad5d7c3f76f2cd3 : SUCCESS in 1h 28m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eeae0f0d9ed644f99f8d966b1412c0a9 : SUCCESS in 1h 12m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/425983c5bac5427ca0c3c0d8a1374665 : SUCCESS in 3m 32s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/3af9cefb50eb42799fd193a00705932a : SUCCESS in 1h 04m 24s","accounts_in_message":[],"_revision_number":2},{"id":"98f79fdcfcaa3fe541a2d7aa5324debf6fe233ff","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-08-30 21:11:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"764e6873d76bb3fb30e39dccd26cd2fea279ab64","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-03 17:05:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"50e01f5647d20acf20c680ace0d1df6b1b36a58e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 17:05:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"723a7d2d9b6bead3ffff3d1f2285103f750d1add","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 19:27:08.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/88495405f5a243ffbdc6ad6080a0ce3d : SUCCESS in 7m 18s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/3c5020080b1945ebbf144bde253aec6c : SUCCESS in 1h 25m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/268461234a664faaa7b0b44554398c74 : SUCCESS in 2h 14m 19s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/224a93931bc940708a43f790467ed607 : FAILURE in 43m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5892e6199e9f4717a7340aa00e8fa904 : SUCCESS in 1h 29m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b53ef5a1095b429184bb030e3914f8b8 : SUCCESS in 3m 48s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/68523d36fbc949a698910a5dac8bd80c : SUCCESS in 1h 25m 45s","accounts_in_message":[],"_revision_number":2},{"id":"0eff236442764b159ff7cbb2935b0cf5238754f7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-03 20:00:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"00462d605a9b321f5b72671463b8a85708186144","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 22:52:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/09402af7b4984bdd869885701a4e2542 : SUCCESS in 5m 14s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/8270721ea1154ea7843bb9b3792868f1 : SUCCESS in 1h 20m 26s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9e094bdd4e294c34a0ee9cfa0f5c8d9c : SUCCESS in 2h 42m 35s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9f9fbaf9451648c781d1d639897c269e : FAILURE in 41m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09c4930ff53c46b487a9d766a56c3238 : SUCCESS in 1h 21m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9177b2fc6e2c45d5adf1e29416254098 : SUCCESS in 3m 50s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/dd7ff80c91cf4972814be7155f15e062 : SUCCESS in 1h 53m 50s","accounts_in_message":[],"_revision_number":2},{"id":"f61fa0f9aa508e349d1f4bd21b01c9c8f4154f3c","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-09-04 12:11:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5c58b0f9a281c6a26f85026330282db58ff98047","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 14:29:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8879b080ff414509982abc0a6964057a : SUCCESS in 7m 09s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/0f8b854074a54f90b614a226e9eccda3 : SUCCESS in 1h 18m 58s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d30b05ea20f546c78321ff138c8fb574 : SUCCESS in 2h 14m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/92789b66a614458382cc01091ecf0c32 : RETRY_LIMIT in 4m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcd7c325147244f090e1480cf1918a10 : SUCCESS in 1h 17m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c206c578d411435ea65770cd09b095cd : SUCCESS in 4m 17s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/6bbc7e3d0e854f92af6b22ca18acecb0 : SUCCESS in 1h 22m 59s","accounts_in_message":[],"_revision_number":2},{"id":"c48903ca8fff3fc75d89d8ae62f0191644ace2fa","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-05 06:03:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7f2856fa9f891d8a106b9c729b3a8872e88a1d85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 08:58:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/70fef1a725b94c47845863a14bedea69 : SUCCESS in 5m 16s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/cb574719281a4a04b2cb3c33f4ab8167 : SUCCESS in 1h 09m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d6a86d11313b4f5685811338ee03ec17 : SUCCESS in 1h 44m 35s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ea30b9d1d4514475aa5007f5e9ff9072 : FAILURE in 43m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d87e9820b13448aaab9bca6f6f07d1d8 : SUCCESS in 1h 21m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da9586820894421897ac66283f8801d6 : SUCCESS in 3m 50s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/ea6f28c73d8c408abe1bbebcfe93d546 : SUCCESS in 1h 22m 36s","accounts_in_message":[],"_revision_number":2},{"id":"6b5036e93bde5a4de32bae4f480bba03e25aae2e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-05 14:23:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cc629b76e9c3822dc284f755d89bd8e7942bb1bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 18:03:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2db986ec86194eea92a9874f6e007ff4 : SUCCESS in 6m 34s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/79f20ba37d1540068433e2328b8f8767 : SUCCESS in 1h 32m 43s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/048a0584af4a4bf39af47cd741b7b05a : SUCCESS in 1h 43m 15s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2dfd2be156144811b76b95445b733308 : SUCCESS in 1h 26m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4fd24da3275403fa53b039601704563 : SUCCESS in 1h 04m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71102b49bd2e491ab316036118109c75 : SUCCESS in 3m 42s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/406e35ff3c9b4024b04ed3fae3389759 : SUCCESS in 1h 10m 10s","accounts_in_message":[],"_revision_number":2},{"id":"46262c5c4cb25c0d0849cf9fa3eba278d21fe11e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 18:04:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a60883bc09661fa9ab13ef5d686c1fd2fa82b1bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 20:14:56.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4edc00f820e94e97a839ba84e277b5e7 : SUCCESS in 6m 11s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/2d3de30c16e046419dd7e7e91e3d9477 : POST_FAILURE in 1h 47m 17s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/18f9b09ac6a54b4a82a783d4c4bc3ed3 : SUCCESS in 2h 01m 34s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0d2a4ae07d9a44d7b03e24a0b2091e8c : SUCCESS in 1h 24m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ec40b32f25b4a67a9cd278607ef4419 : SUCCESS in 1h 19m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a62bcf69560440a8b85a03d65d35b710 : SUCCESS in 6m 24s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/89dc0a96cf27450ca980807437f9fe7c : SUCCESS in 1h 24m 38s","accounts_in_message":[],"_revision_number":2},{"id":"a33efd94e6f23e1a47611f3ead97c34712f1e61b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-05 21:48:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3ecbffa45e47c34b0a6963dda9da4bb03ba2b22c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 23:50:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/85985515f3144e0c8558d6dfb47b6793 : SUCCESS in 5m 07s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/00f92c14f2374e1a96d0f1701ef6f832 : SUCCESS in 1h 08m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d657351a5cba4c38a0b96feb34559d61 : SUCCESS in 1h 24m 17s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ba4582723d774bedb18866ae922ba988 : SUCCESS in 1h 26m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ada61f167494d06a6caf774041d1c52 : SUCCESS in 1h 02m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34e4755e643c492b90d81a953307dd2f : SUCCESS in 3m 29s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/09e10b2b22da4f8e883e109c3019c698 : SUCCESS in 1h 21m 42s","accounts_in_message":[],"_revision_number":2},{"id":"730d430f8c2d6a323c60740c8b33679a4d2be9cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 23:50:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"881cbc9a9c2bf35b32fc0e24bf094741c828b482","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 01:42:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0f7a5d7d89df45bd92ab1557172d50f4 : SUCCESS in 9m 00s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/731dbd7743bf42be9077eb8166917998 : SUCCESS in 1h 13m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b8a914ab947b441199116a33e0e5e6ba : SUCCESS in 1h 45m 25s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/005be3039a634ac09cd11b54af1c0993 : SUCCESS in 1h 15m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/421bed1f43084971b5ea3cfa870e1cfa : SUCCESS in 1h 08m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da40428b473640c8982130752bbcd7d0 : SUCCESS in 2m 56s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/d52faec9e5b14263a2095d80ce0b43c5 : SUCCESS in 1h 12m 00s","accounts_in_message":[],"_revision_number":2},{"id":"f970eb5ede2420720a9ed9251d2204770f4c7088","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 01:42:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9db08f242f42ec2178445e5166fbd56487be1313","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 01:43:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c76b284ec5ad4483be40233c732264eb : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2},{"id":"12d09fd6bc9a7c8d136a8afc6c4944c0a4a012d6","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-09-06 11:46:10.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit e0583fd473e424c36bc8bad6aae2e19bc896556d","accounts_in_message":[],"_revision_number":2},{"id":"e6f2c40d2769c6389435b08407762cd431cb1116","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-09-06 11:46:28.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 1b2719870681d587f15562f752abb9cdb124aa35","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"af0907f12f7d0617e1a42c484561707f798b5a36","revisions":{"8de6ab349df42460b24a6ac2704a908a0cf0e560":{"kind":"REWORK","_number":1,"created":"2019-08-29 16:37:29.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/04/679304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/04/679304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/04/679304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/04/679304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/04/679304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/04/679304/1"}}},"commit":{"parents":[{"commit":"b28287a0fbaf7a88c343ee98fe6a2bc5e275e7a5","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b28287a0fbaf7a88c343ee98fe6a2bc5e275e7a5"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-08-29 16:22:00.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-08-29 16:22:00.000000000","tz":-240},"subject":"Enable amphora nova flavor extra_specs","message":"Enable amphora nova flavor extra_specs\n\nto allow fine tune amphora instances\n\nChange-Id: I011662ad85bc3ac2a9f405e5e0df0af1e050544f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8de6ab349df42460b24a6ac2704a908a0cf0e560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8de6ab349df42460b24a6ac2704a908a0cf0e560"}]},"branch":"refs/heads/master"},"af0907f12f7d0617e1a42c484561707f798b5a36":{"kind":"REWORK","_number":2,"created":"2019-08-29 21:28:04.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/04/679304/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/04/679304/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/04/679304/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/04/679304/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/04/679304/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/04/679304/2"}}},"commit":{"parents":[{"commit":"b28287a0fbaf7a88c343ee98fe6a2bc5e275e7a5","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b28287a0fbaf7a88c343ee98fe6a2bc5e275e7a5"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-08-29 16:22:00.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-08-29 21:27:53.000000000","tz":-240},"subject":"Enable amphora nova flavor extra_specs","message":"Enable amphora nova flavor extra_specs\n\nto allow fine tune amphora instances\n\nChange-Id: I011662ad85bc3ac2a9f405e5e0df0af1e050544f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/af0907f12f7d0617e1a42c484561707f798b5a36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/af0907f12f7d0617e1a42c484561707f798b5a36"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
