)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~680671","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~stable%2Fstein~I011662ad85bc3ac2a9f405e5e0df0af1e050544f","project":"openstack/openstack-ansible-os_octavia","branch":"stable/stein","hashtags":[],"change_id":"I011662ad85bc3ac2a9f405e5e0df0af1e050544f","subject":"Enable amphora nova flavor extra_specs","status":"MERGED","created":"2019-09-06 11:46:10.000000000","updated":"2019-09-07 09:19:11.000000000","submitted":"2019-09-07 09:18:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680671-1567847890646-413f103c","meta_rev_id":"e2fffe4d278b8cb126cb4d413606d37567e28896","_number":680671,"virtual_id_number":680671,"owner":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-09-07 09:18:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-09-06 12:06:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-07 07:42:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-07 07:42:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 12:06:09.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-09-07 07:42:11.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-09-07 09:18:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c29341e3caecdac5c50558a9a66944a78d9ede14","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-09-06 11:46:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"86d5fb39572a0318ddf4ced572d0874c22ac9c77","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-06 12:06:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"65c6f76118a2acf6e5ebc26f389e16989a934139","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 13:18:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a3159b92bbf84368af3c810de91ca1e9 : SUCCESS in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e04fde91c7884cee87680a68e2e7f5e3 : SUCCESS in 3m 51s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 https://zuul.opendev.org/t/openstack/build/9836d4b3872f48f8bbc66d5dd4ff6e79 : SUCCESS in 1h 23m 56s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable https://zuul.opendev.org/t/openstack/build/ee221a00b8494d8fb6ed11eb4eb51dca : SUCCESS in 1h 11m 03s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a970cb9e84a42e89d88994cdf8da98f : SUCCESS in 1h 06m 44s","accounts_in_message":[],"_revision_number":1},{"id":"6e92726408189d3e3a8e7659096214bc66dd6e35","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-07 07:42:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e93f9efceed370f7dde84fb7649b960ac76d1878","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 07:42:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"65d4d690a03d96db4bc30d037846c0be777d38c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 09:18:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9e1897c47cda4eefb4b59cf2ff425401 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9214791f3c8c4562b08cf9352f6f0704 : SUCCESS in 2m 47s\n- openstack-ansible-deploy-aio_metal_octavia-centos-7 https://zuul.opendev.org/t/openstack/build/9f291c440b844ca7aee48f5c0ebb3cfa : SUCCESS in 1h 30m 57s\n- openstack-ansible-deploy-aio_metal_octavia-debian-stable https://zuul.opendev.org/t/openstack/build/f8a1c115f9e94bb5adfda7fee6bbe328 : SUCCESS in 1h 11m 40s\n- openstack-ansible-deploy-aio_metal_octavia-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1446f66f79ca462d84ddfc5115d10c3b : SUCCESS in 1h 11m 56s","accounts_in_message":[],"_revision_number":1},{"id":"77e44f0776208b1fa695dc870cf840e047327d70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 09:18:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e2fffe4d278b8cb126cb4d413606d37567e28896","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 09:19:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d621d81a95fc4d2e949b1ad33ef93c9b : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0583fd473e424c36bc8bad6aae2e19bc896556d","revisions":{"e0583fd473e424c36bc8bad6aae2e19bc896556d":{"kind":"REWORK","_number":1,"created":"2019-09-06 11:46:10.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/71/680671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/71/680671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/71/680671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/71/680671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/71/680671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/71/680671/1"}}},"commit":{"parents":[{"commit":"44ea565f2d65d213f5399541d9cb0cb9ba5bf2fa","subject":"Convert dynamic includes to static imports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/44ea565f2d65d213f5399541d9cb0cb9ba5bf2fa"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-08-29 16:22:00.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-09-06 11:46:10.000000000","tz":0},"subject":"Enable amphora nova flavor extra_specs","message":"Enable amphora nova flavor extra_specs\n\nto allow fine tune amphora instances\n\nChange-Id: I011662ad85bc3ac2a9f405e5e0df0af1e050544f\n(cherry picked from commit af0907f12f7d0617e1a42c484561707f798b5a36)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e0583fd473e424c36bc8bad6aae2e19bc896556d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e0583fd473e424c36bc8bad6aae2e19bc896556d"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
