)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~680672","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~stable%2Frocky~I011662ad85bc3ac2a9f405e5e0df0af1e050544f","project":"openstack/openstack-ansible-os_octavia","branch":"stable/rocky","hashtags":[],"change_id":"I011662ad85bc3ac2a9f405e5e0df0af1e050544f","subject":"Enable amphora nova flavor extra_specs","status":"ABANDONED","created":"2019-09-06 11:46:28.000000000","updated":"2021-01-13 18:00:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e57ebe50830f20ef076e6b0cf1a028a92e954e78","_number":680672,"virtual_id_number":680672,"owner":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-05 15:52:09.000000000","_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"date":"2019-10-23 13:46:47.000000000","_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":-1,"date":"2019-09-06 13:10:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-12-06 21:12:02.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"value":2,"date":"2019-09-06 12:06:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 12:06:18.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-09-06 13:10:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-05 15:52:09.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2019-12-06 21:12:02.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"dddf4024de0d375d963a874aeb8ec8866fd36895","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-09-06 11:46:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"79572b7ad4579b06e29904cee9b8509702f5ac43","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-06 12:06:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d320ede42f2a703a3f0e23a0fba2315964fe3133","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 13:10:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d524a1c33a3e43c6b883a004abf9ccde : SUCCESS in 3m 02s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a14ce1dc9a3848a4b7f10f751e56363f : SUCCESS in 8m 03s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/59f233976e3948edb7b36acea6d65c32 : FAILURE in 1h 19m 02s","accounts_in_message":[],"_revision_number":1},{"id":"02b67af6abad1918e12dcb70a0da24b21f3d4fac","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-09-06 17:53:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"70d6e910afe84ebf4ff11d0bcf5d57925bedd7e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 19:24:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3a26ed2bad9e4b36bab769f56bcccf93 : SUCCESS in 3m 44s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/108f75f9c31f4cc780ae07e62c0980f7 : SUCCESS in 6m 22s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/9e6d2c655fe14f929e976db60f33b657 : FAILURE in 1h 24m 36s","accounts_in_message":[],"_revision_number":1},{"id":"eaef5fea53f181d6ec6e34ab8544c0b4d7d3db55","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-07 19:08:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"68b104aaa77130dc9de96ae1886a600fd1261e19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 20:24:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/993be1b293e74050b158d37f0b2a44db : SUCCESS in 4m 10s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bc106d82eb864e0d9c33be56589c73e5 : SUCCESS in 5m 07s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/0d6d02e8f3ab48b9bc51ddab53512a94 : FAILURE in 1h 10m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9f0b3eab71d13bb4ec77221265f1d0900d27a72d","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2019-10-23 13:46:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6db6524c9c52afea8ad4be16cc8ea79a21cb7741","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 15:01:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e43a8bf461d64e26916145c33c7eb195 : SUCCESS in 4m 38s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/63a3820ba8d34c4f9f7dbbf8659a8502 : SUCCESS in 5m 23s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/221c9de2422b4b02b9755f362204ea9d : FAILURE in 1h 07m 14s","accounts_in_message":[],"_revision_number":1},{"id":"4097ac89f67193fc6f23e154f8cae1e20a4c22e2","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-12-04 11:42:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0993b00e5bd8e78d074a7037b79bd303946af571","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 12:55:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8d4e6fc752ec44f79d7dc85803644208 : SUCCESS in 4m 24s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/57c91954381d47dd831724c0d6be9d1e : SUCCESS in 6m 00s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/45086cfb95b142f5b51b362ffddd40bd : FAILURE in 1h 06m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ca1561e969c757be350fa2cc4fa71ab1c676c0f5","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-12-05 15:52:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ae79e5dcb4f82859142aa5884b35e6fa17f281f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 17:13:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8f551323ee2c42beb1ae1df605d8b4b3 : SUCCESS in 5m 10s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e5472fd60c844d97bb658a16ee066e55 : SUCCESS in 7m 45s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/38816a2f16aa4cc2adac2410b8279f6a : FAILURE in 1h 11m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a20c390ddcb1e14ab4b156bcb6c9d9b3c16da6b9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-12-06 21:12:02.000000000","message":"Patch Set 1:\n\nI think this is going to fail because octavia have updated their API version in the test amphora image and it\u0027s not compatible with the Rocky version of the service.","accounts_in_message":[],"_revision_number":1},{"id":"90e382fac5f6f6ad5f2400f9e13000acf4a3b96a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-01-10 21:57:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"25f14d61af59a091a17df7da252d480d96723b29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 23:11:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/91f9d94efdcf473f848f9f820166fb07 : SUCCESS in 4m 32s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6f051a34458543338b7159487d4c6b41 : SUCCESS in 6m 34s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/77075c9bb6124bbda1575d1306bde7c4 : FAILURE in 1h 08m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e57ebe50830f20ef076e6b0cf1a028a92e954e78","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-13 18:00:05.000000000","message":"Abandoned\n\nrocky is extended-maintainance, abandoning","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b2719870681d587f15562f752abb9cdb124aa35","revisions":{"1b2719870681d587f15562f752abb9cdb124aa35":{"kind":"REWORK","_number":1,"created":"2019-09-06 11:46:28.000000000","uploader":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"ref":"refs/changes/72/680672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/72/680672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/72/680672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/72/680672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/72/680672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/72/680672/1"}}},"commit":{"parents":[{"commit":"91f7e6c345fe26c0bb365a75d94de83843af4be1","subject":"Merge \"Ensure that admin endpoint is used to configure octavia quota\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/91f7e6c345fe26c0bb365a75d94de83843af4be1"}]}],"author":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-08-29 16:22:00.000000000","tz":-240},"committer":{"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","date":"2019-09-06 11:46:28.000000000","tz":0},"subject":"Enable amphora nova flavor extra_specs","message":"Enable amphora nova flavor extra_specs\n\nto allow fine tune amphora instances\n\nChange-Id: I011662ad85bc3ac2a9f405e5e0df0af1e050544f\n(cherry picked from commit af0907f12f7d0617e1a42c484561707f798b5a36)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1b2719870681d587f15562f752abb9cdb124aa35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/1b2719870681d587f15562f752abb9cdb124aa35"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
