)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~681592","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~Ib1993911a076a31e421bc5a61e1d61c6ecd650e3","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"osa-py3","hashtags":[],"change_id":"Ib1993911a076a31e421bc5a61e1d61c6ecd650e3","subject":"Allow venv python interpreter to be overridden","status":"MERGED","created":"2019-09-11 20:14:27.000000000","updated":"2019-10-01 13:52:04.000000000","submitted":"2019-10-01 13:49:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681592-1569937753615-01385176","meta_rev_id":"68751b48d5abbf2d1d22a911d7a20d5c694d58b5","_number":681592,"virtual_id_number":681592,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-01 13:49:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2019-10-01 09:42:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2019-10-01 09:42:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":1,"date":"2019-10-01 11:49:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-30 14:33:46.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-10-01 11:49:42.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-10-01 13:49:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"455e1de547078c96df37a2cc446e079aac4842cb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-11 20:14:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dca0fac908f2ceb977bcded9f76cbf7375f371fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 22:02:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3e81ca7fb4194fe8b1277a5a62b126df : FAILURE in 6m 56s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/c825f7e68c7547e686986ad16bcf41e5 : FAILURE in 1h 12m 39s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/111aaa9916d141b684f030aec7d11b73 : FAILURE in 1h 35m 20s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/bb33b65d21e643568d8303cc3066df2d : FAILURE in 1h 24m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf9c21c32c38490285604b4b9c4d666a : FAILURE in 56m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1325db641f847dab8a18428a468b0f0 : SUCCESS in 3m 04s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/34f7c06562854a88bc65f7e87e345679 : SUCCESS in 1h 15m 23s","accounts_in_message":[],"_revision_number":1},{"id":"af6ada101e1b65304fe1683d9083f973c772ee28","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-11 22:20:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"49db00615a292916ef0ac808a13cf8c48159d17e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 00:25:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/830e53b187e147d8bbfbc8cd7b0faabd : FAILURE in 6m 42s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/dfa73628bfa1458291246dfbaf45cf0b : SUCCESS in 1h 25m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fb44bd011ab94059b9dac6a70a0ab822 : SUCCESS in 1h 58m 37s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4cb213ecfb9b4e9b8139a4e37a16c564 : SUCCESS in 1h 28m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0eee4aac10a4e739e3bd733671daf6e : SUCCESS in 1h 21m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38d104834d84491798ef6179c20004dd : SUCCESS in 4m 09s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/e68aebee733f44e9a99b5902bbe7548d : SUCCESS in 1h 48m 56s","accounts_in_message":[],"_revision_number":2},{"id":"0cfcfb2fd9b025f0278bcb20f7a234f8b941026a","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-09-12 02:54:00.000000000","message":"Patch Set 2:\n\nI guess we need to merge this first \nhttps://review.opendev.org/#/c/680389/","accounts_in_message":[],"_revision_number":2},{"id":"afe2eae0fd3a7beb49e5b987677f040c8cbba0cf","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-13 21:46:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cdfedfea8ef66172ecf6180aca95641223d8f4c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 23:45:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f2fbcc2d482943409ccdaf89f1c615e9 : SUCCESS in 6m 28s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/8d3b7a8702ec4bfa815e8742f6f07e73 : SUCCESS in 1h 23m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7e734e3ad214408a916a77d0bfec2682 : SUCCESS in 1h 52m 25s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9331ff60b5544c05b8e6134a045135d4 : SUCCESS in 1h 27m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51ac1b6a6c9f492cbe376a8f1e0d1f98 : SUCCESS in 1h 25m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffe7892ee9a14ebca6342e0ac91f4034 : SUCCESS in 3m 37s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/8ec67df0bee544f990a5960c34e84b5c : FAILURE in 31m 06s","accounts_in_message":[],"_revision_number":2},{"id":"ceb450ec7ca704b551534a55767db3c95b0f2111","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-18 19:19:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"29429d8478810ebc5ac2365664bbdbe5d57058b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 21:27:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e7121688d57e4ac8bd541c0195faebbe : SUCCESS in 5m 33s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/b62af83a92ef4ce7a58009cbded57a36 : SUCCESS in 1h 29m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/79c33549648a42628e70313967435572 : SUCCESS in 2h 04m 40s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ae188c335a9f4367925382b33efb2244 : SUCCESS in 1h 32m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3d3b0b3eb6e4a02b0031ac7334b140b : SUCCESS in 1h 19m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ca1d00703064340ba5dc327ec45417c : SUCCESS in 3m 25s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/17d3118e41204c488c6e6680021ce168 : SUCCESS in 1h 16m 26s","accounts_in_message":[],"_revision_number":2},{"id":"00775b7bebe7968640fad9e99c8ee5bd0a10d3f4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-20 13:47:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b4da3e0bbb2db83c213107ac339916f4f5caa284","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-09-30 14:33:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a8f3489593d39ba4984522e7c35e62a5c17be386","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 14:33:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"03bc54ea303177324df7e1bf002401deb84cd3cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 16:32:02.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6d4d9ed0fe28489e8a1e745d4c0788b3 : SUCCESS in 7m 05s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/2a0e4f21e74341b4bd17de926b56220a : SUCCESS in 1h 44m 00s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4e8a5810d74c45fa8bdf65f0f52cd614 : SUCCESS in 1h 57m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8bc2638689304631beff1885aab3a785 : SUCCESS in 1h 46m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0af0c8e2934147dfb90b55a6a1c102b0 : SUCCESS in 1h 33m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa372f9071304e91adb71d9c3b1f17bf : SUCCESS in 3m 18s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/6fc8440bb4124acb82cce68464e27d08 : FAILURE in 1h 17m 20s","accounts_in_message":[],"_revision_number":2},{"id":"352b3a20a30f58cf56bd32713a93a70c291f2a88","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-01 09:42:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"75c3050ff673dc6c40f58a6586d3003a445e34f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 11:35:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0d24858c0bdb4764bb9ccb0f470173cd : SUCCESS in 6m 21s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/3a760c7a13a14272adf29791549e2355 : SUCCESS in 1h 34m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/849f4a1dd71f4c9297f7d66b819d7a77 : SUCCESS in 1h 49m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b0691e79dc7e43189d35952566179605 : SUCCESS in 1h 39m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fabc6e72bd1f4ef5a625f2bfe903a69b : SUCCESS in 1h 13m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4239350f8c7940fc9fc1a37c1854b7a5 : SUCCESS in 3m 30s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/99381ae70a394bf19e95ad05c0845768 : SUCCESS in 1h 18m 44s","accounts_in_message":[],"_revision_number":3},{"id":"906cc5476c8486bffb43eca05558d6abe0f793ed","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-01 11:49:42.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"eef371ba70f8d255b862e076a62a1af27e1ed698","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 11:50:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a30ae6c71ef03eab0b523472dfb50d6b2be7190a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 13:49:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/54e06b3e3ba442cc80b63a9b24f4fd54 : SUCCESS in 6m 50s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/61350281d122402f8b75c7d5065afc1f : SUCCESS in 1h 27m 45s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/df63760488aa4612aa37d31a6c9c1d79 : SUCCESS in 1h 56m 29s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7b96a91d897b4ff09ea060c96fd89712 : SUCCESS in 1h 36m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e41b3e529f0544c0b1f32d49e95f4248 : SUCCESS in 1h 11m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1950b21b3f0d48e98e2cd463f898325f : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":3},{"id":"967ae6a4f892e50fea9be4061ccd8af72d1fdf09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 13:49:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8f1d7541a55651eb5b235e79274b4ff974d6e729","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 13:52:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f2c909c9934b0c8c70a6059aec3a62 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2e57a199f204ddb83da2d307ad6b13e9ad6e694d","revisions":{"885d01697b07a6be5702e2e006f5531af0f6ddd3":{"kind":"REWORK","_number":1,"created":"2019-09-11 20:14:27.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/92/681592/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/92/681592/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/92/681592/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/92/681592/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/92/681592/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/92/681592/1"}}},"commit":{"parents":[{"commit":"f8f7a48d1df503eddebf1a97ef3a626fc88f180f","subject":"Merge \"Enable amphora nova flavor extra_specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f8f7a48d1df503eddebf1a97ef3a626fc88f180f"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-09-11 20:13:23.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-09-11 20:14:06.000000000","tz":60},"subject":"Allow venv python interpreter to be overridden","message":"Allow venv python interpreter to be overridden\n\nChange-Id: Ib1993911a076a31e421bc5a61e1d61c6ecd650e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/885d01697b07a6be5702e2e006f5531af0f6ddd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/885d01697b07a6be5702e2e006f5531af0f6ddd3"}]},"branch":"refs/heads/master"},"6b07d0a67ac36d7be2840e507ed23fd86e47b6fb":{"kind":"REWORK","_number":2,"created":"2019-09-11 22:20:18.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/92/681592/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/92/681592/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/92/681592/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/92/681592/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/92/681592/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/92/681592/2"}}},"commit":{"parents":[{"commit":"f8f7a48d1df503eddebf1a97ef3a626fc88f180f","subject":"Merge \"Enable amphora nova flavor extra_specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f8f7a48d1df503eddebf1a97ef3a626fc88f180f"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-09-11 20:13:23.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-09-11 22:20:15.000000000","tz":0},"subject":"Allow venv python interpreter to be overridden","message":"Allow venv python interpreter to be overridden\n\nChange-Id: Ib1993911a076a31e421bc5a61e1d61c6ecd650e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/6b07d0a67ac36d7be2840e507ed23fd86e47b6fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/6b07d0a67ac36d7be2840e507ed23fd86e47b6fb"}]},"branch":"refs/heads/master"},"2e57a199f204ddb83da2d307ad6b13e9ad6e694d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-10-01 09:42:47.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/92/681592/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/92/681592/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/92/681592/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/92/681592/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/92/681592/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/92/681592/3"}}},"commit":{"parents":[{"commit":"eff5be4b997ed7a4ed21f5db134d0823555f8ffe","subject":"Merge \"Octavia certificate distribution\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/eff5be4b997ed7a4ed21f5db134d0823555f8ffe"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-09-11 20:13:23.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-10-01 09:42:47.000000000","tz":0},"subject":"Allow venv python interpreter to be overridden","message":"Allow venv python interpreter to be overridden\n\nChange-Id: Ib1993911a076a31e421bc5a61e1d61c6ecd650e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/2e57a199f204ddb83da2d307ad6b13e9ad6e694d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/2e57a199f204ddb83da2d307ad6b13e9ad6e694d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
