)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~681642","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~Ief7f60972c62388cee1f579e8988c4ba57c8fb65","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"mgmt-subnet-gateway","hashtags":[],"change_id":"Ief7f60972c62388cee1f579e8988c4ba57c8fb65","subject":"Create mgmt subnet with gateway if provided","status":"MERGED","created":"2019-09-12 05:50:42.000000000","updated":"2019-09-19 19:52:17.000000000","submitted":"2019-09-19 19:50:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681642-1568922638698-f233d7c7","meta_rev_id":"634754c9f4ffb6043d07e61e909f4018901d49bc","_number":681642,"virtual_id_number":681642,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"date":"2019-09-18 11:28:52.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-09-19 19:50:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"all":[{"value":1,"date":"2019-09-19 07:35:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2019-09-18 19:11:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":2,"date":"2019-09-19 10:09:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-18 18:59:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":1,"date":"2019-09-19 06:34:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":1,"date":"2019-09-19 10:09:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 18:59:27.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-09-18 19:11:11.000000000","updated_by":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"reviewer":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"state":"REVIEWER"},{"updated":"2019-09-19 06:34:23.000000000","updated_by":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"},{"updated":"2019-09-19 07:35:33.000000000","updated_by":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2019-09-19 10:09:47.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-09-19 19:50:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0064f1bf0489d2590f9e65346fe8ce1745680e27","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2019-09-12 05:50:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2db4485647ccee81eb657270b3ad1734de593689","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2019-09-12 06:44:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"7cd2b5a19b452694899ada88fb28fe8548b28611","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 09:09:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b2df57f814514d859541563acfc347b0 : SUCCESS in 7m 23s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/f9dd5ee5e2634c848a34062ef8f27e3d : FAILURE in 1h 40m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7b402794fd1f4a5899fb4f27601397a3 : SUCCESS in 1h 56m 32s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b7320730e56e4c98b636f47715085092 : FAILURE in 1h 29m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7569c0543a3848978962af9691f9df52 : FAILURE in 2h 15m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/392f4736d9834d1faea8a1c6bec27c38 : SUCCESS in 3m 11s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/362ba7c972bf4382aa3833a3d3a7061d : SUCCESS in 1h 25m 10s","accounts_in_message":[],"_revision_number":2},{"id":"bc35c7075afd7b23a7a902a4759a3ca949dcb473","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2019-09-12 09:52:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"924514434942efdd5b119b6e92b3a648bdc182b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 12:26:55.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/04ff834201f342c3925ed2ab1df043f0 : SUCCESS in 5m 52s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/d13bc527a08445d1b5ff2b01957b6ffc : FAILURE in 1h 27m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8187dc63834f43d1bc9abe335535fdb4 : SUCCESS in 1h 47m 07s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/df785af574444c1a995c9593a7738e84 : SUCCESS in 1h 46m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ca0685702204ec080efa04dfb7bdd21 : FAILURE in 1h 19m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61d8b33d380e4e529d8a791826714f6f : SUCCESS in 4m 18s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/79cfeeda10ed49b5be7639ca32e862d4 : SUCCESS in 1h 29m 28s","accounts_in_message":[],"_revision_number":2},{"id":"06a3a7890dea0574d7f704d085c2c314493b92cc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2019-09-12 13:18:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f80d0c3ced724a0b273b3263613b067de50c201f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 15:26:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c063de35c45b45d3a7aebe46caa0526e : SUCCESS in 6m 06s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/66d91b686d58418595169aeed51f5da1 : FAILURE in 1h 39m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/55d5aafef1e548ac8a9a229f92abc3d4 : SUCCESS in 2h 01m 09s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/84515700c7334035b3bb72877cc4f5d8 : SUCCESS in 1h 42m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae7b1ae7fabe46caab2ef1aac9975ccd : FAILURE in 1h 10m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/969c97d044b848049d4918a1a9018143 : SUCCESS in 4m 31s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/cf78df91c65249ea8f6ad2fd1718ecab : FAILURE in 1h 11m 58s","accounts_in_message":[],"_revision_number":2},{"id":"3b8add64c06f84b538fc0c168e104a07acb22fa7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2019-09-12 18:57:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c99fa59126f8cceaaff494768230bf1ec0a49198","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 20:48:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7944d6ab9fb34e17aaf77f3c2881940c : SUCCESS in 4m 38s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/e50dcc84c9094587934f1aa7f38cd463 : SUCCESS in 1h 36m 45s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/09058c398b354078b3af5402e532c4ff : SUCCESS in 1h 45m 07s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/742e76a2697f48e597b3a0d0ae6d04c5 : SUCCESS in 1h 42m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d075d4ed2b24d7db34b3288e788a441 : SUCCESS in 1h 42m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96348f3487654ce2883d702c4e2b7f9d : SUCCESS in 3m 14s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/1dd850764d1645c3b9ffb351ec648782 : SUCCESS in 1h 40m 13s","accounts_in_message":[],"_revision_number":2},{"id":"d427b00be5a04e8970774ec6a560d0513704d76b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-12 20:55:39.000000000","message":"Patch Set 2: Code-Review-1\n\nCould you please add this new variable (commented out if necessary) to defaults/main.yml with a breif description. That file is used as the primary documentation reference for the role variables.\n\nSee https://opendev.org/openstack/openstack-ansible-os_octavia/src/branch/master/defaults/main.yml#L329-L348","accounts_in_message":[],"_revision_number":2},{"id":"f551fa3fc7ed42769df995facb902c46b17d9ad1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-13 10:47:19.000000000","message":"Patch Set 2:\n\n\u003e Could you please add this new variable (commented out if necessary)\n \u003e to defaults/main.yml with a breif description. That file is used as\n \u003e the primary documentation reference for the role variables.\n \u003e \n \u003e See https://opendev.org/openstack/openstack-ansible-os_octavia/src/branch/master/defaults/main.yml#L329-L348\n\nAlso a release note for the new variable addition","accounts_in_message":[],"_revision_number":2},{"id":"d366ea81310ce944d439a79104118924bc63ef77","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2019-09-14 15:19:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"394d466429a64d8bb63db4b8616079bde9704252","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 17:09:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fa77ce41499243d18a598b394014bb70 : SUCCESS in 8m 12s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/32830c24bc6a4e66b972aaeefe00a9ba : SUCCESS in 1h 17m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dde043b65f524c2691567d05bf09dc47 : SUCCESS in 1h 43m 18s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/70e36ccdc7d64721869fad251e902e62 : SUCCESS in 1h 34m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b87faeb70ffa4ac1ac79c4f496fa8f19 : SUCCESS in 1h 32m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03300136ca794f0d93f2568b593f2d37 : SUCCESS in 2m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e028ffb0f6a147c786229afb3b1613a4 : SUCCESS in 3m 04s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/938cab4a527e4a0f9bbfb400b29cbcdc : FAILURE in 40m 51s","accounts_in_message":[],"_revision_number":3},{"id":"a767ec7614a61c7de3c2e8c28e01692bbc5d8e60","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2019-09-15 07:03:56.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2160a7748a2abf9673d1a3be27e4ba57f4396af1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 08:56:43.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/414f16c5efe14c9f8a15b605e245c026 : SUCCESS in 6m 35s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/955f90f4ce694db5ab5477e7a4dc5bf8 : SUCCESS in 1h 36m 24s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/afd55fe41c8d498db6b519b59c967899 : SUCCESS in 1h 49m 56s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/43e88b9a412d44d09d828fc9648fee94 : SUCCESS in 1h 38m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f38c655df3764a63ad4799085846372d : SUCCESS in 1h 21m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c71568728944f99b74b8d569b4efd97 : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44f118fd3d6343fe9a16d8fa8293a92c : SUCCESS in 3m 37s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/6cd1b5c096e54eadbf2c45d19111a6c0 : FAILURE in 40m 36s","accounts_in_message":[],"_revision_number":3},{"id":"db5d7b4438bcdda1f505d86611d406050a97c4bc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2019-09-15 09:02:54.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"ba8ff368503a65a30d6afd60564740fe3475a6a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 11:00:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1a6ddda5e7ce4ae4ade10476ce2682d1 : SUCCESS in 6m 56s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/e253f13ff85a4f419206634d9e4c5a61 : SUCCESS in 1h 27m 43s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5be06cdab42f4f88a4c93db3a78815b1 : SUCCESS in 1h 46m 09s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/044e6523a0084c18a44cb31228d9686d : SUCCESS in 1h 41m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a46721765f24e9a8d236667f9b8e1eb : SUCCESS in 1h 24m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79ad5924fc654d23bfbb4659205cc4af : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55382fd09bf845b89ce9135e9eab5993 : SUCCESS in 3m 45s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/56dc77e88baa4a7eb4245f8eaf5abbeb : FAILURE in 30m 35s","accounts_in_message":[],"_revision_number":4},{"id":"d73a73cd106c978a5abb33069cb2438ce47faeab","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2019-09-15 12:37:02.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d91700c93d8ac2cf8db7b563393f3c2e1b9ff307","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 14:35:24.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6c7b5076b7bb48f4aa9f30c547de16f4 : SUCCESS in 5m 19s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/a80c976fe949432db62fbc46fab631c9 : SUCCESS in 1h 37m 58s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7b56474f3c62437386a373950bf8c6bc : SUCCESS in 1h 52m 32s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/be7e0eca467e4719a3aa9c268e685e79 : SUCCESS in 1h 33m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/381def894a194e568745d640c9dc8f6d : SUCCESS in 1h 12m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b7fd16c4ff746839ea5bd75758da407 : SUCCESS in 3m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a71527e622344fabc52b4f7b7bc635b : SUCCESS in 3m 09s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/6caed63238f74e5ca90817f7db059bce : FAILURE in 28m 48s","accounts_in_message":[],"_revision_number":4},{"id":"1898e02381c48c46c19c1bc567357a7d845712ee","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2019-09-15 21:17:17.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e2e74e26f3413039382bb7292d7a319221dbb684","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 23:02:11.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3f212c3d2282456085254164dbce5de4 : SUCCESS in 4m 25s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/66dfe600492740c5b150d70ae4be80f2 : SUCCESS in 1h 25m 30s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c9d3df0b94ba4e0895698adaaa07d133 : SUCCESS in 1h 43m 24s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/806374567f8c43c6bad03d0e68c210ae : SUCCESS in 1h 41m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92104b5318c5421e842637e87efcf8b3 : SUCCESS in 1h 23m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/957556ef060f4534ba77eaf0fd1c3da3 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/190ae72827274ca4ae2cfefac26fbc48 : SUCCESS in 2m 49s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/80844083fade450ab81fe59be24d124c : FAILURE in 32m 22s","accounts_in_message":[],"_revision_number":4},{"id":"0c7f3382d2558b29415f4ed22b8c8b0fb8aa4e3c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2019-09-16 06:55:21.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"78eaaad3ba88df1545fefb0553ec5dfe43947a52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 08:59:50.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b2b33fd5369e4750be858a99470e6136 : SUCCESS in 9m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/a9e2e8de88c14da888238ac81f6837a4 : SUCCESS in 1h 35m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ca1aad1cb87c402aa41f111e2a0ce8e2 : SUCCESS in 1h 49m 15s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/01e3defaa6f84539b524f71adcf08eee : SUCCESS in 2h 02m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0629d83675294fb9859c10b097fa989d : SUCCESS in 1h 33m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79bd3f39b03a49f5a6b075acdd1a07e2 : SUCCESS in 2m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7911c5e9319a4183aeec6d333b15db10 : SUCCESS in 5m 12s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/d715cdb0cfda4ae7b0abb5601980623a : FAILURE in 30m 32s","accounts_in_message":[],"_revision_number":4},{"id":"0c7122d1457f36925dbe8843b5ce42f3e02598e6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2019-09-16 17:53:47.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e27f74c796c269b0529037bd2044d88d554236d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 19:59:12.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fa20ad593c4f4375bd0e9d94bdefe3ac : SUCCESS in 7m 23s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/c230497069f94a1498752b45eb9942e8 : SUCCESS in 1h 35m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c453d86bcb8f4408b3b97e60d9f7ffeb : SUCCESS in 1h 59m 38s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4782448401c64a7195e72ccc2f1e1aa9 : SUCCESS in 1h 50m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a22aa3e156f6413cb7e7ff689d078aa9 : SUCCESS in 1h 18m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95e67e3cae6d4187b5f29b5d4195a0e2 : SUCCESS in 3m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62d88cd3463a4b14a1858f46d7d2a7c3 : SUCCESS in 2m 51s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/44a525206dd64c3badbfc51070b01908 : FAILURE in 30m 05s","accounts_in_message":[],"_revision_number":4},{"id":"015719d729a0d0aa36f6572ab01c367c949a5f68","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2019-09-18 11:28:52.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2566e6c8c801ff395622fa67625f50446d404c04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 13:44:30.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fb3345220f2f408eb97dbc673a32b87c : SUCCESS in 5m 27s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/a7f92886b33142a799f8df61e6feda3f : SUCCESS in 1h 40m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e4c6f05784e64a76a4756dc1eb88b6da : SUCCESS in 1h 54m 04s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a3ae359cf32f4b619df8692888b666ee : SUCCESS in 1h 34m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/605e7bb94a7b4456ba782098e34d6e7f : SUCCESS in 1h 11m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b4e125b02bc437cbf87993c6575f1e4 : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ee57a9b14874c7596833e03b2bff6e9 : SUCCESS in 4m 04s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/40cc27948a264d85b76bee63b5ddfe86 : FAILURE in 27m 00s","accounts_in_message":[],"_revision_number":4},{"id":"4f1a4c9e56cf3f679f33c3b4284d126172846b95","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-18 16:26:03.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"7bd17ea7d1e60a3aa9d02ace87a144f7172b8646","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 18:27:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9d1d49a9cb2c497c96e171b1385e5ca8 : SUCCESS in 6m 01s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/47530b65f2484099894147a474870f98 : SUCCESS in 1h 24m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c957238aaefd44a5b612dfb03fcb4687 : SUCCESS in 1h 59m 02s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/fc5628fafc6a4b0aab80c8052efffb83 : SUCCESS in 1h 23m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3862c8fb1b5f4f9a9abb0181fd963885 : SUCCESS in 1h 12m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bb534c118064cd58a58ecc081bc1c51 : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f68080d5e3cf49789e5af0e2a0277938 : SUCCESS in 3m 24s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/199aeffbba2149a3a0914ef3e6440598 : SUCCESS in 1h 29m 44s","accounts_in_message":[],"_revision_number":4},{"id":"98a67b5cfa571648b86e38e5caa82597e1d7e52d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-18 18:59:27.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2454e28974385dd040bcd6621247e69375712491","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2019-09-18 19:11:11.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"301e385df57739d870634c401e59418661239d1d","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2019-09-19 06:34:23.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"58935023306c558e7fa57f007fd4e686137d0e7e","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2019-09-19 07:35:33.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"64e063ca398a3f7c3b84a79243ce386149976e27","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-19 10:09:47.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1482ca882463632a773430d84f7d5c9b3d245459","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 10:10:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6bf5787fc679ddaa80a4fd56dcb76083ab0c637c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 13:35:53.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/46dd29672c5b473f96726739ade60bd1 : SUCCESS in 5m 40s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/400a4d92279344d4a30cd740c14761f8 : SUCCESS in 1h 16m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9e2de33c80e740bdb49df6942f2a5cf9 : SUCCESS in 1h 47m 38s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/db1fa5e9a60b418f8deb2089f6ffea17 : SUCCESS in 1h 38m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac17ce4843be4835a9a6ca5cd0fe64d1 : FAILURE in 42m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8214abfa04ac4ea7a1175a04713c0fb8 : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c36f5f0cc5249d3975c5efe6c56e155 : SUCCESS in 2m 57s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/16940399e6384f3983ef18f9c6cb3d22 : FAILURE in 1h 21m 58s","accounts_in_message":[],"_revision_number":4},{"id":"b8a1e004cb1eb44cbe135381ce9ede48fb77d387","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-19 15:34:08.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f07b5153392c45b483f0f5ddaae138ca1b4cb7a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 17:28:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/eb320eae90d14cf8aedcba113a56bbfd : SUCCESS in 5m 25s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/2f3a0bb7ccee4f15b484a43e25faa980 : SUCCESS in 1h 19m 10s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1ecd4f8c9bde48d998164c36ad9a8706 : SUCCESS in 1h 48m 06s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e35fe67af55b4c03b613ccb25e4d55ad : SUCCESS in 1h 38m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81d80f1e47ad47b787a25a8cf7f88c14 : SUCCESS in 1h 13m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e4b15d1db1f48d38ec805bb651e8d5d : SUCCESS in 3m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f26441f8c7c64c45b6d100cbd4ce9216 : SUCCESS in 4m 49s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/515c7d892e38473a93cab7c63e4c3ea3 : SUCCESS in 1h 44m 52s","accounts_in_message":[],"_revision_number":4},{"id":"f01338b8b61f110a0b473a481627c4213c52cc7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 17:28:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"88983176bac9b8b44c55264fdafe9182d04a6dfc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 19:50:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0b32c0c786574da5ac40f2e3d313c219 : SUCCESS in 5m 27s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/fcc94c8808124731a64835f454a049c6 : SUCCESS in 1h 33m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/567841106c2e434da8833cb10548abfd : SUCCESS in 2h 15m 00s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7d59be7ead1d4f3db6a871fd9dc23e15 : SUCCESS in 1h 30m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05b4ed2701af4705aafc5b5de06a73c7 : SUCCESS in 1h 17m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a8c60818f094808af36139ad395db81 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3600b8a578f4418b987699277e2ac4b : SUCCESS in 3m 23s\n- openstack-ansible-deploy-aio_distro_metal_octaviav2-centos-7 https://zuul.opendev.org/t/openstack/build/674f09c5c813445b934f325f8d6ce306 : SUCCESS in 1h 27m 29s","accounts_in_message":[],"_revision_number":4},{"id":"dfb0dca173742821da0c10dd376b0253e484cd30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 19:50:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ac799f35f5bf5f75a7390e4445ae63c133b2f56e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 19:52:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a89dac8bf5c34b85906e080c0f814dfa : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c20227e38b3427d9550d7ea94150842 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d1380ea1676fdbbc55b85ff3992afac8e85da05e","revisions":{"ed76188ae4c0515b9c8b40dd556b55e32bc4f085":{"kind":"REWORK","_number":1,"created":"2019-09-12 05:50:42.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/42/681642/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/42/681642/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/42/681642/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/42/681642/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/42/681642/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/42/681642/1"}}},"commit":{"parents":[{"commit":"f8f7a48d1df503eddebf1a97ef3a626fc88f180f","subject":"Merge \"Enable amphora nova flavor extra_specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f8f7a48d1df503eddebf1a97ef3a626fc88f180f"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2019-09-12 05:43:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2019-09-12 05:43:38.000000000","tz":180},"subject":"Create mgmt subnet with gateway if provided","message":"Create mgmt subnet with gateway if provided\n\nThis change adds the ability to create the mgmt subnet with default\ngateway when the specified variable provided.\n\nChange-Id: Ief7f60972c62388cee1f579e8988c4ba57c8fb65\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ed76188ae4c0515b9c8b40dd556b55e32bc4f085"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ed76188ae4c0515b9c8b40dd556b55e32bc4f085"}]},"branch":"refs/heads/master"},"0e86e3876a426e01f4d7af4600a67c1863bf1b97":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-12 06:44:21.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/42/681642/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/42/681642/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/42/681642/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/42/681642/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/42/681642/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/42/681642/2"}}},"commit":{"parents":[{"commit":"333e6f1fc7699aeef30f4e9d63c7ede79777c763","subject":"Start using uWSGI role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/333e6f1fc7699aeef30f4e9d63c7ede79777c763"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2019-09-12 05:43:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2019-09-12 06:44:21.000000000","tz":0},"subject":"Create mgmt subnet with gateway if provided","message":"Create mgmt subnet with gateway if provided\n\nThis change adds the ability to create the mgmt subnet with default\ngateway when the specified variable provided.\n\nChange-Id: Ief7f60972c62388cee1f579e8988c4ba57c8fb65\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/0e86e3876a426e01f4d7af4600a67c1863bf1b97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/0e86e3876a426e01f4d7af4600a67c1863bf1b97"}]},"branch":"refs/heads/master"},"af960503ee3c2d29a3beceb86c514e5b45644c90":{"kind":"REWORK","_number":3,"created":"2019-09-14 15:19:42.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/42/681642/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/42/681642/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/42/681642/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/42/681642/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/42/681642/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/42/681642/3"}}},"commit":{"parents":[{"commit":"333e6f1fc7699aeef30f4e9d63c7ede79777c763","subject":"Start using uWSGI role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/333e6f1fc7699aeef30f4e9d63c7ede79777c763"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2019-09-12 05:43:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2019-09-14 15:19:17.000000000","tz":180},"subject":"Create mgmt subnet with gateway if provided","message":"Create mgmt subnet with gateway if provided\n\nThis change adds the ability to create the mgmt subnet with default\ngateway when the specified variable provided.\n\nChange-Id: Ief7f60972c62388cee1f579e8988c4ba57c8fb65\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/af960503ee3c2d29a3beceb86c514e5b45644c90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/af960503ee3c2d29a3beceb86c514e5b45644c90"}]},"branch":"refs/heads/master"},"d1380ea1676fdbbc55b85ff3992afac8e85da05e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-09-15 09:02:54.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/42/681642/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/42/681642/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/42/681642/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/42/681642/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/42/681642/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/42/681642/4"}}},"commit":{"parents":[{"commit":"7a8e052276009bc99e106e9c8d41287fc6ea70e8","subject":"Merge \"Start using uWSGI role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7a8e052276009bc99e106e9c8d41287fc6ea70e8"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2019-09-12 05:43:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2019-09-15 09:02:54.000000000","tz":0},"subject":"Create mgmt subnet with gateway if provided","message":"Create mgmt subnet with gateway if provided\n\nThis change adds the ability to create the mgmt subnet with default\ngateway when the specified variable provided.\n\nChange-Id: Ief7f60972c62388cee1f579e8988c4ba57c8fb65\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/d1380ea1676fdbbc55b85ff3992afac8e85da05e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/d1380ea1676fdbbc55b85ff3992afac8e85da05e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
