)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~690368","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~stable%2Ftrain~If22140803c27cdf78480339511ae8131f5a4ffc0","project":"openstack/openstack-ansible-os_octavia","branch":"stable/train","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"If22140803c27cdf78480339511ae8131f5a4ffc0","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-10-22 19:20:50.000000000","updated":"2020-04-03 09:53:23.000000000","submitted":"2020-04-03 09:50:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690368-1585907415915-87f092a5","meta_rev_id":"38e3bb6ef01e38f6602ad7fcab90ad1cac03d85f","_number":690368,"virtual_id_number":690368,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-03 09:50:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-03 07:39:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-03 07:39:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 07:39:26.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-04-03 09:50:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"30d6cdee39d1fef6d0e929699808092292f28457","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-22 19:20:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0baa840b6db7e02b823d3bfa89c2558244c81140","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 19:30:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e8d9e7a53c6049adb7e29e3dbcc39687 : SUCCESS in 4m 20s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/92c1b7e96b35481ca785b27c76c7cd37 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d377dcf3026470daa8cb8cf087b697d : FAILURE in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d9c3dc79d064c67ad8ec1e443bc0a3c : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"a19939979c352475409cd413e526e86d28bdc8e5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-24 18:35:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7fc3a22243f367c9ec96851742bb2f66d78d5207","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 21:04:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6e85011b83844736abc4c19b42802767 : SUCCESS in 10m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/af0f5d3aff7847308ae4608c05494cbb : SUCCESS in 1h 14m 09s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/3354a4223c77465bb911696673b4f880 : SUCCESS in 1h 38m 00s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5461c52bc5004b309c00a1bcdc5e509d : RETRY_LIMIT in 4m 54s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/283c912713054bc4bf46fc961fe77a7c : SUCCESS in 1h 30m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4fd7448557f84166ad40a1168562a1cd : SUCCESS in 1h 19m 07s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/042c3f063b7e408fbe3789fe42a4273c : FAILURE in 1h 29m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b37aeea38f8e4c59ae7e56fccdcc7f88 : SUCCESS in 2h 02m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/708c1c67638c4ff093be5b1a826bb457 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5106adfe6c4649bb95237d58819cde39 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":2},{"id":"83f728af8c833268cfe7a39b03fb50f3f9a30f65","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-02 20:32:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e1d85b047b71d2217f39bde671c9f5bd11e86d5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 23:52:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/faeaba83deaf47f29d99690d6ffefc40 : SUCCESS in 8m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4d912879f2b9471ea3894e318106a60b : SUCCESS in 1h 31m 51s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/fd0b3367cbc44b2fbe6a93128091ecd3 : SUCCESS in 1h 28m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f26f345847134c7ab9fccebe21a14b72 : SUCCESS in 1h 47m 03s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/111279a668da4d6eb56637eb1fc8ef1b : SUCCESS in 1h 27m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/984642e5df664fa9995c91a7f34276c6 : SUCCESS in 1h 17m 22s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c49fa58528cd40a985b9c232283eab75 : SUCCESS in 2h 55m 39s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c9d5a3fd4194064b22c4959c025ebaa : SUCCESS in 2h 07m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcde6a7e939045f2a0959e31aee574bf : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b5a7fb72cdc4c139bd20a76a4d3994a : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":2},{"id":"f9ca9d66302c5b2ed8ee9e3f05dedcd1e8203f9d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-03 07:39:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a5d20a5d66881a949248002a373734ee9f470dde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 07:39:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7ef97856246c62780eb3c136ed4d34c7804503e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 09:50:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c585cb813352595ec71cc8461d821e5f1d5e8e26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 09:50:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b7be1aed72b44cdd91ba32162df9a0db : SUCCESS in 10m 55s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f60d7c7567cd469abb256ccb79a9fae5 : SUCCESS in 1h 14m 11s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/f01b1b3dbdcd483db9c522a218f1a6fc : SUCCESS in 1h 50m 42s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a582421220ca4ba592283b9300672b1e : SUCCESS in 1h 44m 35s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/acd64c2938014b738f04087a42df6b18 : SUCCESS in 1h 34m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34fde889886b448dbfe3ec6285efd27b : SUCCESS in 1h 30m 04s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ac3fe25b0dc47d9924db4048a2abb2d : SUCCESS in 2h 04m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c330b3bf0f9d45309e3858f8c5c19040 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81506e051c574aeb80e0cc01c6783937 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":2},{"id":"f49a238252d3da0161deff5dba7952a6a2072f99","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 09:53:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a54010d33494ff084ab4939abf2c304 : SUCCESS in 2m 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0e9facac6bd4c4393d132f0358b5998 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7cbec4b075e1af1747b7c7909b57cea2efbcf6b6","revisions":{"0d5212f9c40646a7f2125ba3011649a0946f8539":{"kind":"REWORK","_number":1,"created":"2019-10-22 19:20:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/68/690368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/68/690368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/68/690368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/68/690368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/68/690368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/68/690368/1"}}},"commit":{"parents":[{"commit":"942a245af6cea7de2c056ad1fe6178b61a812908","subject":"PDF Documentation Build tox target","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/942a245af6cea7de2c056ad1fe6178b61a812908"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-22 19:20:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-22 19:20:46.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: If22140803c27cdf78480339511ae8131f5a4ffc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/0d5212f9c40646a7f2125ba3011649a0946f8539"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/0d5212f9c40646a7f2125ba3011649a0946f8539"}]},"branch":"refs/heads/stable/train"},"7cbec4b075e1af1747b7c7909b57cea2efbcf6b6":{"kind":"REWORK","_number":2,"created":"2020-03-24 18:35:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/68/690368/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/68/690368/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/68/690368/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/68/690368/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/68/690368/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/68/690368/2"}}},"commit":{"parents":[{"commit":"61eb9edeb25243059462a999b428baa18fe782fe","subject":"Remove broken experimental job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/61eb9edeb25243059462a999b428baa18fe782fe"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-24 18:35:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-24 18:35:51.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: If22140803c27cdf78480339511ae8131f5a4ffc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7cbec4b075e1af1747b7c7909b57cea2efbcf6b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7cbec4b075e1af1747b7c7909b57cea2efbcf6b6"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
