)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~718635","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I84300d787610ed185b6755f625eea91135ce9cde","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"osa-collections","hashtags":[],"change_id":"I84300d787610ed185b6755f625eea91135ce9cde","subject":"Use ansible openstack collection","status":"MERGED","created":"2020-04-09 09:35:00.000000000","updated":"2020-06-08 20:56:19.000000000","submitted":"2020-06-08 20:27:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718635-1591648063885-bcf263fa","meta_rev_id":"6a699091d537fab3ff7ec236e56c386dbed0691f","_number":718635,"virtual_id_number":718635,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-08 20:27:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2020-06-08 19:06:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2020-06-03 10:28:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2020-06-08 19:06:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 10:28:10.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-06-08 19:06:38.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-06-08 20:27:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"85f591ed77f37797b920cad836654aa42c7abecb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-09 09:35:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75a21b855b1f46682e1fdcd5442f2454ab0e4ee4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-09 09:55:04.000000000","message":"Topic set to osa-collections","accounts_in_message":[],"_revision_number":1},{"id":"588b9656b065302ccda37613d83cf3c1f5ae4906","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-09 10:58:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8cbd2e1e4a83e4454132f12d36ab426a36b0308a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-09 11:44:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a072ef4e6c9dcb92c3b9d84d66b27a9166da1432","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 14:00:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1377e88bfb4e4e2ab5ad810e2ee319a1 : SUCCESS in 11m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/138a56a22b6f491a934c6d5fc3f3169c : FAILURE in 1h 16m 10s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8cb20e529382439d8231add51aefb1ae : FAILURE in 1h 30m 17s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/bdc971dc5a994bbdbe54fac5b6577b9a : FAILURE in 1h 16m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5091f52092b043e89f7093a671f0affb : FAILURE in 1h 04m 06s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d8bc11fe7db24e74ac94a2027f4d4681 : FAILURE in 2h 04m 44s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b41c643494eb42429ea63e79917ce192 : FAILURE in 1h 25m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0579ce28e83643c7aa4fbc79fb8e13ae : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":3},{"id":"f98d4f6da5d693a68fb998047014f45fbb77ca43","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-09 14:20:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"af9c404a4969b67355be8a07cfd7caf84cefecf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 16:35:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d788bd26f30247e9b75a3c836e65e841 : SUCCESS in 1h 11m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5ef2ad4d846b43629f4637a44078e366 : SUCCESS in 1h 57m 56s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f1f0e4d45c474ee897979136870c5885 : SUCCESS in 1h 39m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5189d2df26684e53ba197deaebf5f795 : FAILURE in 49m 16s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6117bb5480c44b018eb74a54a66a5aba : FAILURE in 2h 06m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/547b741599864d1980bc5888eb044a02 : FAILURE in 39m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43e21d2a04bb4ae2aca2d6d60f308666 : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":4},{"id":"355653c7df2fd9902f2624595e767573a8816e4b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-20 12:55:02.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0a2dc7d7fe0c424371e1be611c7901414f6bef3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 16:08:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/465d4b43170f41e98baeafb2a011f9e2 : SUCCESS in 8m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9874b106ff45424f8152d9fd7af1fa03 : SUCCESS in 1h 09m 43s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/33473321f5bc4fff9b5cf04a00bd398b : SUCCESS in 2h 00m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/449b90ecbf434a6f839a087a9392b084 : SUCCESS in 1h 17m 49s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2bed82a6b0ec44af9f5a01c6b0a9d37a : TIMED_OUT in 3h 04m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d10c1bcbe85340d5bca990ddf141ccbc : SUCCESS in 2h 14m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff9a1848d81644068998228c429923ed : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":5},{"id":"896e0712ce1f552c94c0c8619d1fe6dcd932b1a5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-03 10:28:10.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"d4c6d1b98313ecd679be959b093eb7c954d8cded","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-08 19:06:38.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"cad5f4288ce8c33605e47adc5b66dd528d1af2b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 19:06:50.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"729c681e4cc9e57ebaf693022c8661c7a9baca96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 20:27:43.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1e704b85b1ab401594ea689b886e3b18 : SUCCESS in 8m 25s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4d6b212bc09e4f6ea90dc467c8efd939 : SUCCESS in 56m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/173527d6f9e949ae8b45c9a96c2e11ee : SUCCESS in 1h 16m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1eafe4c537d477fb1aca65972440e17 : SUCCESS in 1h 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b63f50cdd8147d39b1c245e284d62d5 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":5},{"id":"947799eb4602d7b2bf5e8e4d917d7cb2947601bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 20:27:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"790995312dd19fe530de2683059b1113639fd375","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 20:30:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a35b2fe3f583439aa351d27bd91eca85 : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":5},{"id":"6a699091d537fab3ff7ec236e56c386dbed0691f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-08 20:56:19.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 501db7ea7a262eaa5554dddcd0932601fc03b51e","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4da762dd13af4bd1a99541766d4fa98146322e14","revisions":{"8081d87dc65c8e53655adda682017c841d96a07a":{"kind":"REWORK","_number":1,"created":"2020-04-09 09:35:00.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/35/718635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/35/718635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/35/718635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/35/718635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/35/718635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/35/718635/1"}}},"commit":{"parents":[{"commit":"a429d60b65f3036ae615b622459ad433ddf14ea2","subject":"Missing document start \"---\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a429d60b65f3036ae615b622459ad433ddf14ea2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 09:34:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 09:34:55.000000000","tz":60},"subject":"Use ansible openstack collection","message":"Use ansible openstack collection\n\nDepends-On: https://review.opendev.org/718431\nDepends-On: https://review.opendev.org/718362\nChange-Id: I84300d787610ed185b6755f625eea91135ce9cde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8081d87dc65c8e53655adda682017c841d96a07a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8081d87dc65c8e53655adda682017c841d96a07a"}]},"branch":"refs/heads/master"},"a37a34dbebcb21bd4c4c4a641515b8900cd33b03":{"kind":"REWORK","_number":2,"created":"2020-04-09 10:58:08.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/35/718635/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/35/718635/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/35/718635/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/35/718635/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/35/718635/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/35/718635/2"}}},"commit":{"parents":[{"commit":"a429d60b65f3036ae615b622459ad433ddf14ea2","subject":"Missing document start \"---\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a429d60b65f3036ae615b622459ad433ddf14ea2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 09:34:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 10:57:52.000000000","tz":60},"subject":"Use ansible openstack collection","message":"Use ansible openstack collection\n\nDepends-On: https://review.opendev.org/718431\nDepends-On: https://review.opendev.org/718362\nChange-Id: I84300d787610ed185b6755f625eea91135ce9cde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a37a34dbebcb21bd4c4c4a641515b8900cd33b03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a37a34dbebcb21bd4c4c4a641515b8900cd33b03"}]},"branch":"refs/heads/master"},"4792e166a7a778ece848f9375db8e8fce3c331d7":{"kind":"REWORK","_number":3,"created":"2020-04-09 11:44:58.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/35/718635/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/35/718635/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/35/718635/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/35/718635/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/35/718635/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/35/718635/3"}}},"commit":{"parents":[{"commit":"a429d60b65f3036ae615b622459ad433ddf14ea2","subject":"Missing document start \"---\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a429d60b65f3036ae615b622459ad433ddf14ea2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 09:34:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 11:44:39.000000000","tz":60},"subject":"Use ansible openstack collection","message":"Use ansible openstack collection\n\nDepends-On: https://review.opendev.org/718431\nDepends-On: https://review.opendev.org/718362\nChange-Id: I84300d787610ed185b6755f625eea91135ce9cde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4792e166a7a778ece848f9375db8e8fce3c331d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4792e166a7a778ece848f9375db8e8fce3c331d7"}]},"branch":"refs/heads/master"},"542c527ea27ad551585bb823fe475c38e72b4d20":{"kind":"REWORK","_number":4,"created":"2020-04-09 14:20:38.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/35/718635/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/35/718635/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/35/718635/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/35/718635/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/35/718635/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/35/718635/4"}}},"commit":{"parents":[{"commit":"a429d60b65f3036ae615b622459ad433ddf14ea2","subject":"Missing document start \"---\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a429d60b65f3036ae615b622459ad433ddf14ea2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 09:34:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 14:20:27.000000000","tz":60},"subject":"Use ansible openstack collection","message":"Use ansible openstack collection\n\nDepends-On: https://review.opendev.org/718431\nDepends-On: https://review.opendev.org/718362\nChange-Id: I84300d787610ed185b6755f625eea91135ce9cde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/542c527ea27ad551585bb823fe475c38e72b4d20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/542c527ea27ad551585bb823fe475c38e72b4d20"}]},"branch":"refs/heads/master"},"4da762dd13af4bd1a99541766d4fa98146322e14":{"kind":"REWORK","_number":5,"created":"2020-05-20 12:55:02.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/35/718635/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/35/718635/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/35/718635/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/35/718635/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/35/718635/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/35/718635/5"}}},"commit":{"parents":[{"commit":"d739992a99a2c6279cbb5db0ce637d3e267ba85c","subject":"Merge \"[ussuri][goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/d739992a99a2c6279cbb5db0ce637d3e267ba85c"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 09:34:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-20 12:54:38.000000000","tz":60},"subject":"Use ansible openstack collection","message":"Use ansible openstack collection\n\nDepends-On: https://review.opendev.org/718431\nDepends-On: https://review.opendev.org/718362\nChange-Id: I84300d787610ed185b6755f625eea91135ce9cde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4da762dd13af4bd1a99541766d4fa98146322e14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4da762dd13af4bd1a99541766d4fa98146322e14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
