)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~754411","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~Icf411ab2df0f6e937ce056b085fee97b89fe8361","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"run_condition","hashtags":[],"change_id":"Icf411ab2df0f6e937ce056b085fee97b89fe8361","subject":"Define condition for the first play host one time","status":"MERGED","created":"2020-09-25 14:47:31.000000000","updated":"2020-12-04 23:02:28.000000000","submitted":"2020-12-04 23:01:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754411-run_condition","meta_rev_id":"5cef52c14c60abdba263ed3f17d4436e29a62a46","_number":754411,"virtual_id_number":754411,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-04 23:01:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-12-04 23:02:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-04 23:02:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2020-12-04 13:44:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-02 11:51:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-02 22:50:41.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-12-03 11:50:43.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"b90f4d29e97c34c6e3bebc56f2fa505cead52872","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-25 14:47:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"457bd692435f637973cde31dc7c01c6a3535bad3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 17:05:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9bb95ce20d0d499db3f845506239f375 : SUCCESS in 9m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/39f66795c55145d399da35dee0863d1a : SUCCESS in 1h 20m 29s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/16c8d05830b94e398614a4b058023f52 : SUCCESS in 1h 22m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be4cd0f92cfa46daa8a3337e3216dbd5 : SUCCESS in 1h 18m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1401596243cc46e3b5f25b3bd5006db7 : SUCCESS in 1h 14m 00s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8bda0262574d4f9ca70f1c8524f33f55 : SUCCESS in 2h 10m 04s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef8fccdabc5d4243885e5208855baf10 : FAILURE in 2h 00m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2662f0c535c64461b058704630ff1309 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"4ce1d38d40ac0660c7273a759e988455e72eea02","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-29 16:56:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8c632cbb6cece4ac268ab89def275426813a9b8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 19:35:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/696eadddd49f4fb6af2cfcae88af94b6 : SUCCESS in 8m 47s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/03e8faebce724bd19d8e66e1b570bd05 : FAILURE in 1h 16m 36s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/af57eb499ccd4bdbb29ec4b951f7346e : FAILURE in 30m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9695a6b3caba4c479c6f7afd86353dcf : FAILURE in 1h 13m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56945b83141c4b7a970ed3701a685dac : FAILURE in 1h 11m 33s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0b8f2ebcfce54f8cafbe5abd041978c0 : FAILURE in 2h 17m 00s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/232f923d8d704505a2da57908ada9058 : FAILURE in 1h 50m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/684547e2e5904b6888f995631f8c43b3 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"26ba2db3d8ba6fa9f3222e714238a1c69c0b694b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-02 09:24:07.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9ccdbaa9af40fa1e3a107c0fa9af335219542e18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 11:51:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/df6a07bf97ff4d628e2aca64b48858ae : SUCCESS in 7m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/38dc1b21bb8445ba8344b400ec61c6f3 : SUCCESS in 1h 17m 47s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/37be35ba144041e68e0a451915a70906 : SUCCESS in 1h 36m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d8fc5dfa4f549e8b13fafe2e4cf5e97 : FAILURE in 1h 05m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47df0d49cecf42ef970798d2dddd1e3f : SUCCESS in 1h 14m 12s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/92863c0989724a2cb99c6071184b0856 : FAILURE in 2h 12m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8952cf30447d48feaee72b81ef80a438 : FAILURE in 2h 04m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adfbf569dcd541aaaa037baf7e13d4e9 : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":2},{"id":"47e0be06a39268b5c8cc295c6e67918e755225f1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-06 14:35:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2f0adfb4f3149dafef579b74e8c8b2f88429252d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 17:11:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6e91562ea1e549e797af626ccec49025 : SUCCESS in 11m 34s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b8c3772583f04bc6b7bd6bd83364e906 : SUCCESS in 1h 14m 03s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a8c407c79bab4542b949885ef7f876db : SUCCESS in 1h 28m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1a05bd20d3548b28da848380617722d : SUCCESS in 1h 16m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52750bfd88dd4d02b88cb09f48458336 : SUCCESS in 1h 13m 45s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2577e188172045388111a61686d14184 : FAILURE in 2h 08m 10s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60c0ae061e4e40b692836513b8949c31 : FAILURE in 1h 49m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2ea1ea24c1b43f5a4f9924d5e45475b : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":2},{"id":"de4cd1c28dbd48064a17ec07dfffdc8f5aa5aabd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-26 10:26:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d07ce4409a27ddc2b6c348f428ba73f09e34c774","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 12:50:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/791e5a4da749432dab545dc7ea7ae054 : SUCCESS in 9m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/de85f340b5974046bf606243009b1396 : FAILURE in 1h 29m 11s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/27099e71f2ba449a89a049fa49095257 : SUCCESS in 1h 47m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/303ee23988584e3eb80647da3609725a : SUCCESS in 1h 22m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa85313c7ca64521bdf2c726dc92fa21 : SUCCESS in 1h 10m 39s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/24abf33688b0485c8e84d35d73221a36 : FAILURE in 2h 14m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdea3af40eb44088a10ffe1a1934837c : FAILURE in 1h 37m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e610b1a709b44972bda96fc719ac884d : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":2},{"id":"1c00582a6ba213728197c6146b014a3d606ac016","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-02 07:54:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"72b227902faf4bf93377b58c7b4aa1f90e58898b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-02 09:30:19.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"14a1a5d9c7724d12d5d19aebd1cbb4a56ce4d6d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 11:06:28.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2cccfc3bf97843679ae0b853c13f259a : SUCCESS in 8m 41s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/070777ccf32948a0b66945fc807500e5 : FAILURE in 1h 15m 46s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/556348de035b4aea9c6a361c13b73997 : SUCCESS in 1h 30m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb1dacc888a944b4b0b559e74a154213 : FAILURE in 30m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3c6f3947a404101a25ed04efd5090fe : SUCCESS in 1h 25m 15s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e419497103ff48b087f6b5d7a91cc7f2 : SUCCESS in 2h 16m 55s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2848a75996a5435cb2ed62d11081dfc6 : SUCCESS in 1h 52m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb661a025cd74283a638e89aeb392618 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":3},{"id":"b0e4c54575103ebe96edbefb0c689ec29dd53eb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-02 17:04:33.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"dd11d5c606c9db98c56f3f19ab7fd78d19f6888a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 20:09:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cc6fa68ce92a4c0d9565b0ae21a14142 : SUCCESS in 8m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/aa3367432f2b479996fb92d945983398 : SUCCESS in 1h 17m 48s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/421c639919c44fbc93ae7a24c5a2ef5d : SUCCESS in 2h 50m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58e10a6706e747bcb1fa407570bb3a3b : SUCCESS in 1h 24m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f8430b1cbc04914a8da05034281b21d : SUCCESS in 1h 57m 27s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f0e45aa3a78b44518c9fcfd2c2a9694b : SUCCESS in 2h 50m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7539fde5e1a649b98b399ea94a1730b0 : SUCCESS in 1h 53m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebe6f048717745a89c5e802688928435 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":4},{"id":"88b5b2dbb0addac37c8f6df6e5b3dde3121504b7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-12-02 22:50:41.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"fb5e2bc848d3f8ec793162c45274999a340f489f","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-12-03 11:50:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"00fccee20a347fe653c215a3471822d6569f480e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 16:05:26.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ebd4f2d2708a76783ff302b5db55078e8ae75e96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-03 18:48:43.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"f51395a6fcdf98aaba5e2565f9f6914fe4312dda","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-03 18:48:58.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"0fba6f6dbfc049c261a16afba1a31a0c2584b1b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 21:29:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4a5da51334b84016973faf956dda60ab : SUCCESS in 8m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7b936cada93e4d2ea1aa692bed63100d : FAILURE in 1h 54m 25s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c46cddc397994eba8c65a52eb8502da2 : SUCCESS in 1h 33m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e62a9e994ab1451a93fc1cb483da6850 : SUCCESS in 1h 28m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1166e739642e46c0b63e0b5527b32b95 : SUCCESS in 1h 27m 27s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/48a3996cf62d48c0978d71474adfbd4c : SUCCESS in 2h 27m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12fffb19ee174651903c0d94e5301c79 : SUCCESS in 1h 59m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8a8c1928f7f40879dea8f513fbf05fd : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":5},{"id":"f6f2885954d53e3ebf11f9bf3c66e53f497e790d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-04 08:09:21.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"a8b3c4944e4cef33ce32ff803f3a0d95196f4d80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 10:39:31.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8fabe1130b6a42cc9fd46295df888e59 : SUCCESS in 11m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b986c9a2fd4a4d899afb56544edddbc9 : FAILURE in 1h 27m 06s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4f2044e3b9c94c918c0defef3877377c : SUCCESS in 1h 43m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6670ed8a35ae4c09a3f0022a54552ce8 : SUCCESS in 1h 25m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/006344d7103549eaa09d34c6f9625911 : SUCCESS in 1h 17m 05s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/505cfb04b2cc481c85fd16e2b0d545f5 : SUCCESS in 2h 22m 13s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3999b1f6a4c40498c10a8abfb8bb81a : SUCCESS in 1h 54m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9ec8b1e382f434b9ea46b5a5ed5ea15 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":5},{"id":"9efc789e440e5baf2a2539369996b6ac91a64813","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-04 13:43:33.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"222a48d94972bdd6ea6570dbc3237a68838ed534","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-04 13:43:59.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"a93b1ce23e36337059c7672d81bf924e79c36415","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-04 13:44:06.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"6509835b66de627ea820df7ae9d9f9d9c4cb9d7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 16:24:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d87f8074f0914e51a2d483984f8ac601 : SUCCESS in 8m 20s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/41e61f621baf4f54a0d4ebe036b879c9 : SUCCESS in 1h 25m 25s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0cf7bb25084649b0b7b7780d0f88f6fa : SUCCESS in 1h 39m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08280a950eae472d82e1031040e0c99a : SUCCESS in 1h 28m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa80a3c15db44cd3af4b5288c672fc3f : SUCCESS in 1h 35m 47s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a4da24c3ca0949629d43eaf1b82832c3 : SUCCESS in 2h 11m 13s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d637f0a9389f48cd935c104f4a0ca448 : SUCCESS in 2h 26m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c974b9b181644c1dbbee25bc902942c2 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":6},{"id":"26c318c8bed9f6181df24ad109caf586e8dd10d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 16:32:54.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"dc3d677c689a39fc0804554752adde89ac72a459","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 18:17:44.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bba2d024cb494a958d8d468fc52b2ee3 : SUCCESS in 7m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c33bbf5ef17f49f4816bd933e1d451d4 : FAILURE in 1h 14m 50s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f912ed0da7b94879b90a68abee22db4e : SUCCESS in 1h 30m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c352fa7efe834f778d21fe0902dcb649 : SUCCESS in 1h 28m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adcbe7396ee340c8bb39c50b937c95a7 : SUCCESS in 1h 16m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c989766c34924c20a87ececc93c15856 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":6},{"id":"7158adac71746f7e81b1a30c78a31a20de703f86","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-04 18:43:28.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"dcc7fc9af27ee4afe9e28c579796be44322b3ae4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 21:16:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d87432b0cdc64b798342a0a4d5214e2e : SUCCESS in 10m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8150ee2bd6f54c6591ab23f3163c3912 : SUCCESS in 1h 21m 53s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/950bbef69b934b4b9d09036c43a8fd21 : SUCCESS in 1h 39m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/faee57c4cca54385863e4fb9428a4a16 : SUCCESS in 1h 19m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f62055d6217477d9ab6541e4efc1981 : SUCCESS in 1h 36m 46s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/98e1111a338d4ebf96141c29476f114a : SUCCESS in 2h 24m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7c64841166a4ff694da492b10e22419 : SUCCESS in 1h 57m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3200c1da8b6f46cd9ae824a7d37c65c4 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":6},{"id":"4a3a85a710b0ff0bbc89cd13c399f4b71a717ac8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 21:16:17.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ee96712b14837ff4ed9fbb171c65d583a1654c23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 23:01:08.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/94aea2defa5743159b2923afc1f1e083 : SUCCESS in 12m 25s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e518e754017b468c939720651128f78e : SUCCESS in 1h 23m 01s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/722cd86e3e2f4ee6922ecf15b189e212 : SUCCESS in 1h 36m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b6d87bfe2b543588835970517a1ff31 : SUCCESS in 1h 24m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4552b4187ddf40489c7c76234de93b90 : SUCCESS in 1h 19m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9db833e3f9774c788067224e16072bb7 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":6},{"id":"0e541f588f39e4a234bfee335101b78ade5545ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 23:01:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"6dd1b2bb3611f7f7b99d65635e2ca5249f5deb6f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 23:02:28.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09d360219a1248ff8747e74c18596f13 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"aef1b1723adda87f66b644c1f4597efd43a32a43","revisions":{"9e01602f3ac34fef8d4fd819a4744d372a245534":{"kind":"REWORK","_number":1,"created":"2020-09-25 14:47:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/11/754411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/11/754411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/1"}}},"commit":{"parents":[{"commit":"7fb7f1ba753e0c146602839ec5f2a676d93830f5","subject":"Use the utility host for db setup tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7fb7f1ba753e0c146602839ec5f2a676d93830f5"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-25 14:47:22.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-25 14:47:22.000000000","tz":180},"subject":"Define condition for the first play host one time","message":"Define condition for the first play host one time\n\nWe use the same condition, which defines against what host some \"service\"\ntasks should run against, several times. It\u0027s hard to keep it the same\nacross the role and ansible spending additional resources to evaluate\nit each time, so it\u0027s simpler and better for the maintenance to set\na boolean variable which will say for all tasks, that we want to run\nonly against signle host, if they should run or not now.\n\nChange-Id: Icf411ab2df0f6e937ce056b085fee97b89fe8361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9e01602f3ac34fef8d4fd819a4744d372a245534"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9e01602f3ac34fef8d4fd819a4744d372a245534"}]},"branch":"refs/heads/master"},"a532868c33dc9234b6ecc4e9dc01a383f443996f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-10-02 09:24:07.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/11/754411/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/11/754411/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/2"}}},"commit":{"parents":[{"commit":"f814fee16cdca1ce6458ea55c68a77d741460928","subject":"Fix linter errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f814fee16cdca1ce6458ea55c68a77d741460928"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-25 14:47:22.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-10-02 09:24:07.000000000","tz":0},"subject":"Define condition for the first play host one time","message":"Define condition for the first play host one time\n\nWe use the same condition, which defines against what host some \"service\"\ntasks should run against, several times. It\u0027s hard to keep it the same\nacross the role and ansible spending additional resources to evaluate\nit each time, so it\u0027s simpler and better for the maintenance to set\na boolean variable which will say for all tasks, that we want to run\nonly against signle host, if they should run or not now.\n\nChange-Id: Icf411ab2df0f6e937ce056b085fee97b89fe8361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a532868c33dc9234b6ecc4e9dc01a383f443996f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a532868c33dc9234b6ecc4e9dc01a383f443996f"}]},"branch":"refs/heads/master"},"8053115e17c7714d5a462464888a8b70d9a6f2be":{"kind":"NO_CHANGE","_number":3,"created":"2020-12-02 07:54:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/11/754411/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/11/754411/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/3"}}},"commit":{"parents":[{"commit":"a3a27aac7f9eb8790c134d3920cc9d1abd1f2a3a","subject":"Fix linter errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a3a27aac7f9eb8790c134d3920cc9d1abd1f2a3a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-25 14:47:22.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-02 07:54:25.000000000","tz":0},"subject":"Define condition for the first play host one time","message":"Define condition for the first play host one time\n\nWe use the same condition, which defines against what host some \"service\"\ntasks should run against, several times. It\u0027s hard to keep it the same\nacross the role and ansible spending additional resources to evaluate\nit each time, so it\u0027s simpler and better for the maintenance to set\na boolean variable which will say for all tasks, that we want to run\nonly against signle host, if they should run or not now.\n\nChange-Id: Icf411ab2df0f6e937ce056b085fee97b89fe8361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8053115e17c7714d5a462464888a8b70d9a6f2be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8053115e17c7714d5a462464888a8b70d9a6f2be"}]},"branch":"refs/heads/master","description":"Rebase"},"26613c632fb04ab68bf4795dac9879db8a36b5bf":{"kind":"NO_CHANGE","_number":4,"created":"2020-12-02 17:04:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/11/754411/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/11/754411/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/4"}}},"commit":{"parents":[{"commit":"9cf165dfce1c3066bd22cd11de6290813abfece2","subject":"Fix linter errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9cf165dfce1c3066bd22cd11de6290813abfece2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-25 14:47:22.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-02 17:04:33.000000000","tz":0},"subject":"Define condition for the first play host one time","message":"Define condition for the first play host one time\n\nWe use the same condition, which defines against what host some \"service\"\ntasks should run against, several times. It\u0027s hard to keep it the same\nacross the role and ansible spending additional resources to evaluate\nit each time, so it\u0027s simpler and better for the maintenance to set\na boolean variable which will say for all tasks, that we want to run\nonly against signle host, if they should run or not now.\n\nChange-Id: Icf411ab2df0f6e937ce056b085fee97b89fe8361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/26613c632fb04ab68bf4795dac9879db8a36b5bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/26613c632fb04ab68bf4795dac9879db8a36b5bf"}]},"branch":"refs/heads/master","description":"Rebase"},"7049a36b21c3b7a225b5f4e6a4d02e32773555bc":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-12-03 18:48:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/11/754411/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/11/754411/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/5"}}},"commit":{"parents":[{"commit":"6cd5a241950348940abe6e88b03a246dbf2df663","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/6cd5a241950348940abe6e88b03a246dbf2df663"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-25 14:47:22.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-03 18:48:43.000000000","tz":0},"subject":"Define condition for the first play host one time","message":"Define condition for the first play host one time\n\nWe use the same condition, which defines against what host some \"service\"\ntasks should run against, several times. It\u0027s hard to keep it the same\nacross the role and ansible spending additional resources to evaluate\nit each time, so it\u0027s simpler and better for the maintenance to set\na boolean variable which will say for all tasks, that we want to run\nonly against signle host, if they should run or not now.\n\nChange-Id: Icf411ab2df0f6e937ce056b085fee97b89fe8361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7049a36b21c3b7a225b5f4e6a4d02e32773555bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7049a36b21c3b7a225b5f4e6a4d02e32773555bc"}]},"branch":"refs/heads/master","description":"Rebase"},"aef1b1723adda87f66b644c1f4597efd43a32a43":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-12-04 13:43:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/11/754411/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/11/754411/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/11/754411/6"}}},"commit":{"parents":[{"commit":"f426ffb541066f94904bba6df2479a4d7a04603b","subject":"Merge \"Update octavia messaging options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f426ffb541066f94904bba6df2479a4d7a04603b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-25 14:47:22.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-04 13:43:59.000000000","tz":0},"subject":"Define condition for the first play host one time","message":"Define condition for the first play host one time\n\nWe use the same condition, which defines against what host some \"service\"\ntasks should run against, several times. It\u0027s hard to keep it the same\nacross the role and ansible spending additional resources to evaluate\nit each time, so it\u0027s simpler and better for the maintenance to set\na boolean variable which will say for all tasks, that we want to run\nonly against signle host, if they should run or not now.\n\nChange-Id: Icf411ab2df0f6e937ce056b085fee97b89fe8361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/aef1b1723adda87f66b644c1f4597efd43a32a43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/aef1b1723adda87f66b644c1f4597efd43a32a43"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
