)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~755576","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I9072c5311ab5ff7b76095c20cad802ddf65efedc","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"osa-linters-update","hashtags":[],"change_id":"I9072c5311ab5ff7b76095c20cad802ddf65efedc","subject":"Fix linter errors","status":"MERGED","created":"2020-10-01 15:53:31.000000000","updated":"2020-12-03 18:19:56.000000000","submitted":"2020-12-03 18:14:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755576-osa-linters-update","meta_rev_id":"b8ba9b35529853fab655b6ededc5e6478b4a4216","_number":755576,"virtual_id_number":755576,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-03 18:14:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-12-03 11:49:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":2,"date":"2020-12-02 16:22:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-12-03 11:49:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":1,"date":"2020-12-03 16:04:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-10 14:15:46.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-11-17 16:46:07.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-11-17 19:03:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 11:49:35.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"7981cd4d90a284073d10408472dbfe52ed280d8b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-01 15:53:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb0a3a9ef1950b710f3ecf047df1e8e6ee36351f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 16:07:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/69b512143af84a1d97af4b3e8a89a073 : FAILURE in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48d50fb61adc41a997567fe700c87d7c : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":1},{"id":"96085e88b6db97a69d4f1a5afb32c4c3142f725c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-01 16:26:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b7578c33923af26c9eca0115a534428721790bd7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-01 16:32:39.000000000","message":"Topic set to osa-linters-update","accounts_in_message":[],"_revision_number":2},{"id":"010071f87db9e31fe274f3f933d41339e96c512c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-01 17:00:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9ad28b921de52d1c57c6ebe21d58f31757b2557c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 18:56:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/eb76b445094b4d6394a988346a48e8ca : SUCCESS in 9m 17s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/13dfe53f1a984f309178a760ef62fa72 : FAILURE in 1h 04m 29s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d759b7de9cb144679ac2616b436fd5bd : FAILURE in 1h 17m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d70d3151d435412db3aee8cd2fcf2e3d : FAILURE in 1h 13m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1db6f5a0e0da47a782a14038bacda39a : FAILURE in 1h 04m 07s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/620126f5491b4c54a00fb8cb912d949e : FAILURE in 2h 13m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe4674ff929f4a41b13d645ea6377a8d : FAILURE in 1h 43m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb253d04902e487485cd034ae57f5e07 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":2},{"id":"10410d2cb9dac87629ffa18ee6497c9f0ce6994a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-02 09:23:43.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e9df2d5f17e719bd40ef1bc0b926a3c1417a80ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 11:39:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/55f86480ca1b4355aeff24dfcad8deed : SUCCESS in 7m 38s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8826d8998c4647c187919b13f902839d : SUCCESS in 1h 10m 33s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b335585f0366459ebb8d00d90ad58cd8 : SUCCESS in 1h 36m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3533542c569249de86723901e1987d52 : FAILURE in 1h 10m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/670f7e1c33f54a84872b5702c878d891 : SUCCESS in 1h 16m 44s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5a47836f10ba4c568fc34fe33bfcd42b : FAILURE in 2h 10m 37s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0ed92aabbe043dfbef7e9ddf5485be8 : FAILURE in 1h 50m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1dd1a0184274bdf913344e08b84f0bd : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":3},{"id":"750918786a485ff438d4484ccfa3fc1785be786d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-02 12:06:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f60d43e370688fabbc10aaed5c3c10a541283a30","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-02 12:06:25.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"13a1d023e2fc35b8f72c4feb10b24cb1fe04c62b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 12:08:15.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"8235594a5e0238378bf946f7ac1b7aec5d0461ca","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-02 12:09:19.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"057544d59544502f21e1f64b855e0560907efbf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 14:39:19.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/600e19dbbd1a4c3aa34ab09f0d18d968 : SUCCESS in 10m 42s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9422ce1abb8a4dccb2bb6dde4c559009 : SUCCESS in 1h 09m 42s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5f979e3b697f4d3d9d14f34b27e2640c : SUCCESS in 1h 25m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cfdce546fd0e4e02b47edf607b32de6e : SUCCESS in 1h 32m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4690b26c22f349449d9373e88e5cc4ff : SUCCESS in 1h 13m 43s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ad384275ca14492786a40d1935b85750 : FAILURE in 2h 16m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e239b79d2e54d1b8ad68d26c2b34e2c : FAILURE in 1h 47m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a300e446f9844d9e9af98ff97e32b806 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":3},{"id":"2b98b398428393dc8689a7b926c97e8792ca5ebc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-06 14:35:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"70b172f5cf22d09fe7f45b65b0775a46c0c2713f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 18:06:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2dced4973bd44f9e960ac31ecba0ee9e : SUCCESS in 10m 53s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0f555c1d0db64ec7b57e03ba5be9776e : SUCCESS in 1h 17m 44s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a3f5e7d558404b6cad59943908a97b3d : SUCCESS in 1h 31m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e3ede3407764a82b8d3e151baab924e : SUCCESS in 1h 22m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a93d1998f77c428c807aff5fd8f1232a : SUCCESS in 1h 54m 55s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/574be743a63e4db88b12b2f2bf29b7cc : FAILURE in 2h 47m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f22082320f634b1396f968a67de00ce7 : FAILURE in 1h 56m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cf69d37143146e2a6286c3bc92513fc : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":3},{"id":"5166924087e20847582abdafe67cd0da3dc51704","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-19 17:19:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"bc97caaf50c04776581d8aa64c136f01d85a1296","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 19:57:59.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5eb80ff921f64325b2dd5954a3ddccf1 : SUCCESS in 13m 11s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bbc4374c62cf41528289a05b5fa88cb4 : SUCCESS in 1h 26m 40s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6e1c8a2f34964dbca687bb3675f249cd : SUCCESS in 1h 43m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0daa6e9595d2431d91dddf244a5adb6c : RETRY_LIMIT in 19m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ac34538f22a4424a4d2fca2ee6d989f : SUCCESS in 1h 10m 31s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/24c718b4dae243edb7093d62446dbb27 : FAILURE in 2h 19m 26s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f442de68ad014ac0a4f19b1335cd8321 : FAILURE in 1h 47m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fd403bd49de4311a7f0b5bbed4f798f : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":3},{"id":"06c4b60b323935bd840b8c18a019015bd49f76f4","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-11-10 14:15:46.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f5c51b09186f9de016d81f768982ee14dc783fe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 16:58:51.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5ae71c897a2d4d5682fbab1757ea055a : SUCCESS in 9m 20s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/dd81b2fb8bd344969e9f185bfc7e95ec : FAILURE in 1h 22m 00s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/87fd440ab7bc457da5f88211b7bc8648 : SUCCESS in 1h 47m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b288dbbf97144ed1a7eeab78bace22e0 : SUCCESS in 1h 25m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f7f2a02ec07480684b7a74b682f5a4e : SUCCESS in 1h 21m 16s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e8a2cdfcb6bb48838ad87686be0ad659 : FAILURE in 2h 08m 35s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11f14e501bf74b639bc6f68f7a41c63b : FAILURE in 2h 30m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc619057e9c3416a95f8cc8795e6f550 : SUCCESS in 17m 58s","accounts_in_message":[],"_revision_number":3},{"id":"d6e7820b5dd709584879918bfd6a6ad09ca3dafe","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-11 09:17:39.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"a3a463d87158882908ba05279d04aeb8fc14e62c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-11 09:17:44.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"369313bacc4a3dc184580ed01d4ac786d83586b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 11:28:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d212797d654f46f69ee1b779f707eae9 : SUCCESS in 12m 58s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/30fdf09dc0eb4a4d974cb3ca158708e0 : FAILURE in 1h 21m 31s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1fc5a3c67148480297c15ff2d3ba19a5 : SUCCESS in 1h 38m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9411bd5809641c6ba66a6b226394f57 : SUCCESS in 1h 31m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1511bc7c0ed943a09597e6db60d53f64 : SUCCESS in 1h 17m 54s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a977b21a4f3f4e30b469dc2845d56b8d : FAILURE in 2h 00m 44s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3e6a58048f944aea338979c65626418 : FAILURE in 1h 49m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab7842c2a61748db86ef25a5afdb312c : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":4},{"id":"882427cae3a44d8a9a090690a17c57eb393118d9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-17 16:46:02.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"14d021f75af1100dc43766f76c2576454f9be69a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-17 16:46:07.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"1bb0ac5c027db9d0c332faf1ebb52784b87ef194","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 19:03:41.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fd11a01242da4206b5118844ecbe3c2d : SUCCESS in 9m 34s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/dcc6e49e52024b7a94fe307afefe5ada : SUCCESS in 1h 26m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/159f7c0a19ef488ca5c7df0120b1f22c : SUCCESS in 1h 41m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27505a87a6ba40abbb43c556bac5eef4 : SUCCESS in 2h 06m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03145234709543ec8b9fb7a37a2c86e6 : SUCCESS in 1h 16m 27s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/53ee78adbe9547c9960eb7b5fcfe51c3 : FAILURE in 2h 11m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed416a2ed3b14f99b3f221f4dfc39bc0 : FAILURE in 1h 42m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d4c33c844d940d883f14b98c819c8b7 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":5},{"id":"d1fc0b311aefd4da4095910a81b2bfe20e199e7d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-27 11:05:22.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"9f652f8f34a6336b45d631998ccd4d9139b51008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 13:37:43.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/353facd271c442ccab84cee5682fe144 : SUCCESS in 12m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/406190955e3641cd9fab4b801edf07f6 : FAILURE in 1h 26m 59s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2c8cedebf4694d2b9366fdf94b5ce8fb : SUCCESS in 1h 39m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f70df0c94f6540a8bc747c70438bd07c : SUCCESS in 1h 26m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cf1140135574c489a8a31047e096116 : SUCCESS in 1h 25m 52s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fc349412f355492db90da34514754a47 : SUCCESS in 2h 18m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9b490d059a84d80a3b7ac0e7273c6f7 : SUCCESS in 1h 51m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95b621d650b04d8c9c74261cbc8fa3cc : SUCCESS in 13m 18s","accounts_in_message":[],"_revision_number":5},{"id":"c06754914c17d6bd9db343fce756e1001d74b9b7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-27 13:38:23.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"64a7d745b53e209ca7497894a13aa4b5b0092f18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 15:59:24.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2d33d580bc0345d6a3f6d2331b7451f6 : SUCCESS in 11m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/aaa8b22ecce646559bcd49a77988b754 : FAILURE in 20m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/cea1e10adeb74363bd3ee39659640ecc : SUCCESS in 2h 12m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9790301b2e854cc990c4db1c8a8f8efc : SUCCESS in 2h 16m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2c03f749d0643b1bc81fee006401b12 : FAILURE in 17m 49s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6e36c1f3b6ba4540a122597a89530950 : FAILURE in 21m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7cc7ff3eedf49aaaaf9ad8eb331f853 : FAILURE in 19m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfe9d893e71943219ee54e98f31e9b0f : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":5},{"id":"8ee05a156451d94163e78b157cb05cd13353d771","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-29 07:52:43.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"8685977a8c5d62b72b16268bf13c17c6480aec5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-29 10:25:17.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7a80e404c9fc420dbb278ef26740639d : SUCCESS in 9m 09s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ee182b4325ac45359a3ddfb6268e481e : FAILURE in 1h 18m 56s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f29a0ccd5208415fb8b2be7f3f565bd9 : SUCCESS in 1h 43m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d441c3d578754660947542def527360b : SUCCESS in 1h 28m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c07dca8e564454a8babf9c56cebb4fd : SUCCESS in 1h 20m 28s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f793b0bce7f6450fad5f50a43de708d2 : SUCCESS in 2h 23m 44s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fd49bf0933c47719bdbf67fa46ce55c : SUCCESS in 1h 52m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a125928f7309460c88e9bb2d89623e2e : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":5},{"id":"0b9b3eec26c6e3049018f2fab951ceb01a89f8ac","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-29 21:59:36.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"62824c2dd4fd0d0ff5d8fa1a3f7b98ecec595305","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 00:10:14.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/832c129547974a548ed01195d208932f : SUCCESS in 11m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b58d8decd7cc493d8e1a06b93b9b261f : FAILURE in 2h 06m 50s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/46560e9b9470430992f94b315c6ec371 : SUCCESS in 1h 46m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0db6196eb32a452f9f2dcda5f541c626 : SUCCESS in 1h 20m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff18df595e83450baec6e3096bc71c2f : SUCCESS in 1h 14m 10s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/557676e6584c41ee9e1770bbfaf1a616 : SUCCESS in 2h 04m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c40548beed2467490d21e8146e7dea5 : SUCCESS in 1h 46m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17e6dba30b934d138f9fd2006b578981 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":5},{"id":"7f5757e34f2d28ef684b47e81a87c06f2ecccc14","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-01 19:13:12.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"2dc8c87f5ea7b780c782a34b3b47f1a0bac8916c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 22:03:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/abaa22274e324e0287da92ace25a9433 : SUCCESS in 8m 25s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d6d68d9d121f4340aa0a84d50dc56a06 : SUCCESS in 1h 13m 51s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/deee71b279fd43eb8f7d60cd599f433b : SUCCESS in 1h 34m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98415546198c479daa67c5d44d44a0f1 : SUCCESS in 1h 23m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff1fda2f95b2411a859bd50db44f16f5 : SUCCESS in 1h 18m 07s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a5685aa101c64cd68c1f6272e7210fce : SUCCESS in 2h 29m 42s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a99ea620e7c45298658df99ea9d748b : SUCCESS in 1h 43m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e375fab090f4dfeb6b9bc097b8caf9b : SUCCESS in 17m 11s","accounts_in_message":[],"_revision_number":5},{"id":"017316787a41b3eb71187c2835d9dba685f299bb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-01 23:06:02.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c1d15096622d27fac46674bb28d892f0f9af48f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 23:06:15.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"841f24c7d2b217b63d47b012e724c9f4010479ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 01:00:37.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0392e902b7a64c469196333bf74833ae : SUCCESS in 8m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/493fb4d9f67a44e19d7547ffac3513f0 : FAILURE in 1h 18m 23s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/af4bcb1d9a5144ebab0d7cd644260969 : SUCCESS in 1h 24m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c24bfdadc8ff4e63937c18ec0144d7a6 : SUCCESS in 1h 44m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4c566237e1c4925baf2fa7c835bfaf1 : SUCCESS in 1h 30m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27d5ce10c62f48c3ab6731521fea2cf4 : SUCCESS in 13m 21s","accounts_in_message":[],"_revision_number":5},{"id":"f9dadbd921116064c9951e8b457820acd0188630","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-02 09:30:45.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"6357d794b0166dbe366c520dd6ea54ac076b533f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 11:58:35.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3ddf8f4c8560491b9787d93cc3c443bf : SUCCESS in 9m 13s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/70ad05d158ee4a23b71cb1cd70855ac2 : SUCCESS in 1h 25m 03s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ecef71675f4a488b83e08ea92d435c75 : SUCCESS in 1h 43m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f573207268074252af6aa4ddef917978 : SUCCESS in 1h 29m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e77b1f0a02d49b89bcc6da1c71e887c : SUCCESS in 1h 30m 06s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6e8781d67f1d46a0b1a789a19c0ee26d : SUCCESS in 2h 13m 43s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d17d2c2047f144d4b00e2b4b0ab73764 : FAILURE in 1h 38m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5885f03abf1c4d268af0981ab2dcd932 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":5},{"id":"553808ca4ffd930280445661450c098dffd77ff8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-02 13:10:54.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"e7b313d659032c64ef451d8ea7ece7d296b0cd6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 16:14:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/28694033991247db856a624ae58cebac : SUCCESS in 10m 06s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/30ac1cefde2d450d91a436851fdb7a33 : SUCCESS in 1h 27m 47s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/eac8335960d5446d95e4ef93d2518f6f : SUCCESS in 1h 43m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32061f2cc4be4644ac4552375944db74 : SUCCESS in 1h 34m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e315e5a868ee49ea93b4492296b5b967 : SUCCESS in 1h 28m 35s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3782224f88ef4c6cbd40c2ab2804f539 : SUCCESS in 2h 42m 30s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3703e4ef743545378344aa8a84e0638e : SUCCESS in 1h 50m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed2d8fd5370843a6b8e03436d26e2d97 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":6},{"id":"217d32860af5d2b3151d30b99209fe9c3bfb49dd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-02 16:22:07.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"d6aff85b35a430c439e86cd5e675baae3b32c967","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-02 22:38:41.000000000","message":"Removed Workflow+1 by Dmitriy Rabotyagov \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":6},{"id":"a71226803d27a0ab1eb1075b0d7cd5f54041cb30","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-12-03 11:49:35.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"d0b03a94e68f9a0597165dfd0eca79cd978bfdcf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-03 16:04:55.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"0f7715df229c4f95bde978a16b4119650ac0b507","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 16:05:09.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8f81ef5662ab16a3b9bfd32e00e3f66282f3d70b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 18:14:17.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bb7b71f6f3f943baa4282b6eb64c5547 : SUCCESS in 10m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9b9c0cf856a74c81b717baf874aa54ff : SUCCESS in 1h 32m 31s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d26c7b6241e541fca61f32af0c9c8054 : SUCCESS in 1h 54m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b6d434a8ddb4597ae270910e3b00de0 : SUCCESS in 1h 31m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cd9977ce549483d91dc7c627095325f : SUCCESS in 1h 25m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/affc01c3d4a34c5789cc0614746e4651 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":6},{"id":"896fa97bcc06e186b81b6fe89a7456841cb57bce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 18:14:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"b8ba9b35529853fab655b6ededc5e6478b4a4216","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 18:19:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c723a841065475ea0e6979c2e78877c : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"9cf165dfce1c3066bd22cd11de6290813abfece2","revisions":{"f48f5f45e49216e485bf9285f1d55f7b11679c1e":{"kind":"REWORK","_number":1,"created":"2020-10-01 15:53:31.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/76/755576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/76/755576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/1"}}},"commit":{"parents":[{"commit":"7fb7f1ba753e0c146602839ec5f2a676d93830f5","subject":"Use the utility host for db setup tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7fb7f1ba753e0c146602839ec5f2a676d93830f5"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-01 15:53:19.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-01 15:53:19.000000000","tz":60},"subject":"Fix linter errors","message":"Fix linter errors\n\nChange-Id: I9072c5311ab5ff7b76095c20cad802ddf65efedc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f48f5f45e49216e485bf9285f1d55f7b11679c1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f48f5f45e49216e485bf9285f1d55f7b11679c1e"}]},"branch":"refs/heads/master"},"d042037f6b668181b224c945b5ddcd0a38fe5c94":{"kind":"REWORK","_number":2,"created":"2020-10-01 16:26:09.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/76/755576/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/76/755576/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/2"}}},"commit":{"parents":[{"commit":"7fb7f1ba753e0c146602839ec5f2a676d93830f5","subject":"Use the utility host for db setup tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7fb7f1ba753e0c146602839ec5f2a676d93830f5"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-01 15:53:19.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-01 16:25:25.000000000","tz":60},"subject":"Fix linter errors","message":"Fix linter errors\n\nDepends-On: https://review.opendev.org/755585\nChange-Id: I9072c5311ab5ff7b76095c20cad802ddf65efedc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/d042037f6b668181b224c945b5ddcd0a38fe5c94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/d042037f6b668181b224c945b5ddcd0a38fe5c94"}]},"branch":"refs/heads/master"},"f814fee16cdca1ce6458ea55c68a77d741460928":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-10-02 09:23:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/76/755576/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/76/755576/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/3"}}},"commit":{"parents":[{"commit":"7fb7f1ba753e0c146602839ec5f2a676d93830f5","subject":"Use the utility host for db setup tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7fb7f1ba753e0c146602839ec5f2a676d93830f5"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-01 15:53:19.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-10-02 09:23:41.000000000","tz":0},"subject":"Fix linter errors","message":"Fix linter errors\n\nDepends-On: https://review.opendev.org/755737\nChange-Id: I9072c5311ab5ff7b76095c20cad802ddf65efedc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f814fee16cdca1ce6458ea55c68a77d741460928"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/f814fee16cdca1ce6458ea55c68a77d741460928"}]},"branch":"refs/heads/master"},"19a22d729b8c3bfbca7e83e094e3609e35c442cf":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-11-11 09:17:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/76/755576/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/76/755576/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/4"}}},"commit":{"parents":[{"commit":"7fb7f1ba753e0c146602839ec5f2a676d93830f5","subject":"Use the utility host for db setup tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7fb7f1ba753e0c146602839ec5f2a676d93830f5"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-01 15:53:19.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-11-11 09:17:35.000000000","tz":0},"subject":"Fix linter errors","message":"Fix linter errors\n\nDepends-On: https://review.opendev.org/758461\nChange-Id: I9072c5311ab5ff7b76095c20cad802ddf65efedc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/19a22d729b8c3bfbca7e83e094e3609e35c442cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/19a22d729b8c3bfbca7e83e094e3609e35c442cf"}]},"branch":"refs/heads/master"},"a3a27aac7f9eb8790c134d3920cc9d1abd1f2a3a":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-11-17 16:46:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/76/755576/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/76/755576/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/5"}}},"commit":{"parents":[{"commit":"7fb7f1ba753e0c146602839ec5f2a676d93830f5","subject":"Use the utility host for db setup tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7fb7f1ba753e0c146602839ec5f2a676d93830f5"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-01 15:53:19.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-11-17 16:46:00.000000000","tz":0},"subject":"Fix linter errors","message":"Fix linter errors\n\nChange-Id: I9072c5311ab5ff7b76095c20cad802ddf65efedc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a3a27aac7f9eb8790c134d3920cc9d1abd1f2a3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/a3a27aac7f9eb8790c134d3920cc9d1abd1f2a3a"}]},"branch":"refs/heads/master"},"9cf165dfce1c3066bd22cd11de6290813abfece2":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-12-02 13:10:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/76/755576/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/76/755576/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/76/755576/6"}}},"commit":{"parents":[{"commit":"7fb7f1ba753e0c146602839ec5f2a676d93830f5","subject":"Use the utility host for db setup tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7fb7f1ba753e0c146602839ec5f2a676d93830f5"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-01 15:53:19.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-02 13:10:48.000000000","tz":0},"subject":"Fix linter errors","message":"Fix linter errors\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/765134\nChange-Id: I9072c5311ab5ff7b76095c20cad802ddf65efedc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9cf165dfce1c3066bd22cd11de6290813abfece2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9cf165dfce1c3066bd22cd11de6290813abfece2"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
