)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~765138","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I1c8f9723ca2ac2b468725c2954adcdaff54dbdf0","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"octavia/api_v1","hashtags":[],"change_id":"I1c8f9723ca2ac2b468725c2954adcdaff54dbdf0","subject":"Drop octavia v1 api options","status":"MERGED","created":"2020-12-02 13:46:43.000000000","updated":"2020-12-03 18:14:46.000000000","submitted":"2020-12-03 18:14:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765138-octavia/api_v1","meta_rev_id":"56909a549f61534c61e5e9ab572ce7014766a32a","_number":765138,"virtual_id_number":765138,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-03 18:14:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-12-03 14:30:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-03 13:22:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-12-03 14:30:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-02 16:38:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 13:22:25.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-12-03 14:30:25.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"61d5422f062f8cd2386149a0c1e967affe9f8177","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-02 13:46:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bd271215f28fc6d592eabead4e57969c6ba1c96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-02 14:27:43.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3a185b14681fe3800c456a24741149a64ef4d7da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 16:38:36.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"cbe04c557c0246a61b83976d41461ba439a6b9dc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-02 21:19:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8ac52de3f9dd444afdbda32fe75dddb5026b6fe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 00:05:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4285a5f5a4c048958b4f3e5c8ea5d078 : SUCCESS in 10m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/12c24bf3d7774fba967c5af8e19cfb41 : FAILURE in 1h 22m 36s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d9c9afdb13294baaa168a551a6b9b326 : SUCCESS in 1h 40m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/637670e1865b44198fc9635a9b269ad9 : SUCCESS in 1h 20m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf24c6d61eaa4d56bf46beda637fc02c : SUCCESS in 1h 57m 00s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/39dedb661dd04109b58f1f35ef20da24 : SUCCESS in 2h 23m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc754c07d0d045d88d9bec0ef51ec380 : SUCCESS in 1h 51m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63da848cd8b04a32a40bb1ce5ea3d333 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":2},{"id":"7e4eb3a4f00cadce1b0fc5cd3e5252d16ec824db","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-03 07:41:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"91bdb702b15488126908d55058d185c4ef2719e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 11:25:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/919db6c570744259a549a8ec6bf8e98c : SUCCESS in 9m 14s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cafeccb2d68148448514df0783665080 : SUCCESS in 1h 43m 01s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/32f19dc4b48544c6b30ae4719cce65d1 : SUCCESS in 1h 24m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf70aa4f5f4d453590f5eecdccc3d20e : SUCCESS in 1h 39m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1851b1c277445208b4d861743aed2c5 : SUCCESS in 1h 17m 20s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6930596876d346f88e6a5997e7816e76 : SUCCESS in 2h 37m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a511e16512fe4b67b2680b11e9130c1c : SUCCESS in 1h 54m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f209edb2f251445daa8f5eaccfebbd7c : SUCCESS in 27m 49s","accounts_in_message":[],"_revision_number":2},{"id":"1a8833aa7357564e7dc69091eee02a5870e56147","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-12-03 13:22:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0ed65eac0d231dd9f4d6c30a9f9ba53c3bbe21ff","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-12-03 14:30:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"346bbf730193c84c19f444fe454b5c2bf408a9ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 16:05:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d56419df8d7932560df0f591a32835ca38e496ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 18:14:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c283d8e0afda483495aeec28763519ec : SUCCESS in 9m 50s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/52ff2f489e194d469d90dfcb78b00a93 : SUCCESS in 1h 21m 35s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ad4900dbee954b87a7765d9b1defb201 : SUCCESS in 1h 43m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/770eb49c5e894af383ef4369c81fb928 : SUCCESS in 1h 37m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6329125a27f742e6b523ef5a9c9a2369 : SUCCESS in 1h 35m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b127e3ccf4a94bce8fc52d69f2318da4 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":2},{"id":"56909a549f61534c61e5e9ab572ce7014766a32a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 18:14:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e00cb9c5636dff40c1e964dd6158919ddc4d8e72","revisions":{"2b109b1dd52df8b3ae8a2a420a0006cb7cfd41ba":{"kind":"REWORK","_number":1,"created":"2020-12-02 13:46:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/38/765138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/38/765138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/765138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/765138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/765138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/765138/1"}}},"commit":{"parents":[{"commit":"7fb7f1ba753e0c146602839ec5f2a676d93830f5","subject":"Use the utility host for db setup tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7fb7f1ba753e0c146602839ec5f2a676d93830f5"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-02 13:46:33.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-02 13:46:33.000000000","tz":120},"subject":"Drop octavia v1 api options","message":"Drop octavia v1 api options\n\nOctavia v1 options have been dropped from upstream on Train. They have\nno effect nowadays so no reason to futher carry it\u0027s codebase\n\nChange-Id: I1c8f9723ca2ac2b468725c2954adcdaff54dbdf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/2b109b1dd52df8b3ae8a2a420a0006cb7cfd41ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/2b109b1dd52df8b3ae8a2a420a0006cb7cfd41ba"}]},"branch":"refs/heads/master"},"e00cb9c5636dff40c1e964dd6158919ddc4d8e72":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-02 14:27:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/38/765138/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/38/765138/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/765138/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/765138/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/765138/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/38/765138/2"}}},"commit":{"parents":[{"commit":"9cf165dfce1c3066bd22cd11de6290813abfece2","subject":"Fix linter errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9cf165dfce1c3066bd22cd11de6290813abfece2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-02 13:46:33.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-02 14:27:43.000000000","tz":0},"subject":"Drop octavia v1 api options","message":"Drop octavia v1 api options\n\nOctavia v1 options have been dropped from upstream on Train. They have\nno effect nowadays so no reason to futher carry it\u0027s codebase\n\nChange-Id: I1c8f9723ca2ac2b468725c2954adcdaff54dbdf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e00cb9c5636dff40c1e964dd6158919ddc4d8e72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e00cb9c5636dff40c1e964dd6158919ddc4d8e72"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
