)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~766693","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I93398ef72be3e423c721f92e4bf9077cd5e08d05","project":"openstack/openstack-ansible-os_octavia","branch":"master","hashtags":[],"change_id":"I93398ef72be3e423c721f92e4bf9077cd5e08d05","subject":"Delegate info gathering to setup host","status":"MERGED","created":"2020-12-11 10:34:59.000000000","updated":"2020-12-14 12:42:39.000000000","submitted":"2020-12-14 12:41:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766693","meta_rev_id":"45050b93b3f18a8f240ed1eab293fbb0a519ad2c","_number":766693,"virtual_id_number":766693,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-14 12:41:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"recommended":{"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},"all":[{"value":1,"date":"2020-12-11 14:17:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},{"value":2,"date":"2020-12-14 10:46:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-14 08:51:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":0,"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},{"value":1,"date":"2020-12-14 10:46:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-11 14:15:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-11 14:17:59.000000000","updated_by":{"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},"reviewer":{"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},"state":"REVIEWER"},{"updated":"2020-12-12 21:00:01.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2020-12-14 08:51:21.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-12-14 10:46:46.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"9e5e89e807a8551ca09ec41bde97df2384d54e17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-11 10:34:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ced66d822751fdfef90dfa95763684320fbd2c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-11 10:44:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0e0bce06441311bd20cf914fab649c442059d5eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-11 12:03:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a88c6824e199beb9c79eadcec1100c21446a98b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 14:15:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d5d106f6557a4054831056900433fab3 : SUCCESS in 9m 21s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cbd4e56c66ae4de4824d41fbc9792d67 : SUCCESS in 1h 33m 51s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6f5ef40c06b149b48f8d653465b35ab5 : FAILURE in 41m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8036f77c37ac404bbe45c96e1728721f : SUCCESS in 1h 35m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbae60d912b64f5a91010c6fcf3b1275 : SUCCESS in 1h 37m 32s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/57b1792780124614b71f2612b551ff1b : FAILURE in 45m 22s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa6a2f4404f547b38cd963865f70465d : SUCCESS in 2h 05m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d20e0aaeb8348f597d9dbdfb080ddf1 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":3},{"id":"c36353708307147fa3ab761b83a58a1bc72c045e","author":{"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},"date":"2020-12-11 14:17:59.000000000","message":"Patch Set 3: Code-Review+1\n\ncode works  +1","accounts_in_message":[],"_revision_number":3},{"id":"888c6e169f949338ad3021723753db9eb2ba050c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-12-12 21:00:01.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6947478aab7e83908dc1c5302520356d206ef5b1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-12 21:47:00.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1bc7a0c62de7b211c6c6fb4625e3679639a9b891","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-12 23:19:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1ae9a863932641279d6c713ce0db1330 : SUCCESS in 7m 47s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/648570a83c804ff2adef596e31387f07 : SUCCESS in 1h 15m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4761f4c590034502890298158974f178 : SUCCESS in 1h 23m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca137c7611fd41c8941518e79c02767d : SUCCESS in 1h 19m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc09d5dd88a74563a1e26650f13aa052 : SUCCESS in 1h 16m 57s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/10fbc19f9fd84e09b14cff44a2c50430 : FAILURE in 38m 25s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ca3ee8970164dbfa116c742deca0ac7 : SUCCESS in 2h 17m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bad815635fa940e69234568058f3752e : SUCCESS in 10m 08s","accounts_in_message":[],"_revision_number":3},{"id":"9047b190dea5d9e78ad92748b857e80ebc824f33","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-12-14 08:51:21.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c95d2f464ba406a5d1d963ceb93db854be11fe72","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-12-14 10:46:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d67cfefd9e826ecc0ffad67e9184554f0a61c469","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 10:46:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"130626d258e99aaf99207e7d72640c3801e5e492","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 12:41:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ec52c79c665b4c56818df16d400a66a0 : SUCCESS in 7m 44s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a904de2385ce46cb94c36e94750495ee : SUCCESS in 1h 19m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6997e75371574f8da6410fa7ab1c3f10 : SUCCESS in 1h 45m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97a8b8ea1417406f9e38a00df0721424 : SUCCESS in 1h 38m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8c472582311456e89a35f091033223d : SUCCESS in 1h 29m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d965672b5e54d88a8341d83d0c554e8 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":3},{"id":"600bac1ac11fa7e1b59fa5a8459464ca5a43d2e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 12:41:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"45050b93b3f18a8f240ed1eab293fbb0a519ad2c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 12:42:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92151471c70647d9a5e422db321c81ed : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7659d1ac44820ab3e467e70e5357200fd85adc09","revisions":{"c2b7516e3023be9ac3688b446b6386c095713b0e":{"kind":"REWORK","_number":1,"created":"2020-12-11 10:34:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/93/766693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/93/766693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/93/766693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/93/766693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/93/766693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/93/766693/1"}}},"commit":{"parents":[{"commit":"b94b927aaf2686cd9311564dfd7173a959f75027","subject":"Merge \"Automatically create and import keypair if it doesn\u0027t exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b94b927aaf2686cd9311564dfd7173a959f75027"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-11 10:34:49.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-11 10:34:49.000000000","tz":120},"subject":"Delegate info gathering to setup host","message":"Delegate info gathering to setup host\n\nWe are not supposed to have neither openstack client nor clouds.yaml file\nin octavia containers and all openstack commands are expected\nto be executed on the setup (utility) host.\n\nChange-Id: I93398ef72be3e423c721f92e4bf9077cd5e08d05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/c2b7516e3023be9ac3688b446b6386c095713b0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/c2b7516e3023be9ac3688b446b6386c095713b0e"}]},"branch":"refs/heads/master"},"5cacd36f40d8ed3321a4cb6185d1a025c459339d":{"kind":"REWORK","_number":2,"created":"2020-12-11 10:44:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/93/766693/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/93/766693/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/93/766693/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/93/766693/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/93/766693/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/93/766693/2"}}},"commit":{"parents":[{"commit":"b94b927aaf2686cd9311564dfd7173a959f75027","subject":"Merge \"Automatically create and import keypair if it doesn\u0027t exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b94b927aaf2686cd9311564dfd7173a959f75027"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-11 10:34:49.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-11 10:44:13.000000000","tz":120},"subject":"Delegate info gathering to setup host","message":"Delegate info gathering to setup host\n\nWe are not supposed to have neither openstack client nor clouds.yaml file\nin octavia containers and all openstack commands are expected\nto be executed on the setup (utility) host.\n\nChange-Id: I93398ef72be3e423c721f92e4bf9077cd5e08d05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/5cacd36f40d8ed3321a4cb6185d1a025c459339d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/5cacd36f40d8ed3321a4cb6185d1a025c459339d"}]},"branch":"refs/heads/master"},"7659d1ac44820ab3e467e70e5357200fd85adc09":{"kind":"REWORK","_number":3,"created":"2020-12-11 12:03:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/93/766693/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/93/766693/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/93/766693/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/93/766693/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/93/766693/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/93/766693/3"}}},"commit":{"parents":[{"commit":"b94b927aaf2686cd9311564dfd7173a959f75027","subject":"Merge \"Automatically create and import keypair if it doesn\u0027t exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b94b927aaf2686cd9311564dfd7173a959f75027"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-11 10:34:49.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-11 12:02:50.000000000","tz":120},"subject":"Delegate info gathering to setup host","message":"Delegate info gathering to setup host\n\nWe are not supposed to have neither openstack client nor clouds.yaml file\nin octavia containers and all openstack commands are expected\nto be executed on the setup (utility) host.\n\nChange-Id: I93398ef72be3e423c721f92e4bf9077cd5e08d05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7659d1ac44820ab3e467e70e5357200fd85adc09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/7659d1ac44820ab3e467e70e5357200fd85adc09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
