)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~768514","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~Ie319fd02cdd60f8a8ac65f0508e9075f40839ae9","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"osa/octavia_barbican","hashtags":[],"change_id":"Ie319fd02cdd60f8a8ac65f0508e9075f40839ae9","subject":"Use barbican for certificates storage","status":"MERGED","created":"2020-12-25 16:09:38.000000000","updated":"2021-03-02 10:14:14.000000000","submitted":"2021-03-02 10:11:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768514-osa/octavia_barbican","meta_rev_id":"75616a153eaf9917422798de56beb9a8fe67f62f","_number":768514,"virtual_id_number":768514,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-02 10:11:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-03-02 08:30:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-02 08:25:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-03-02 08:30:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-25 18:09:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-08 18:23:55.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2021-03-02 08:25:38.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-03-02 08:30:17.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"94a0c0f395bfb321a41eba1af1a0806f4ddbb90f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-25 16:09:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d372ac5513b6d0b88c47a516dd021cfc7de4b30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 18:09:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0420c9216b8c43fab285f10aaf616bbb : SUCCESS in 8m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/67aea84ab9674bb294236c5876c3cf51 : FAILURE in 1h 34m 57s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/91c3e52c3bd34cefaba9a40402bfa822 : FAILURE in 1h 19m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a7be806c1ce40aba2cb29708e481f83 : FAILURE in 1h 22m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8405865fd714eb0af21a906410b8cfb : FAILURE in 1h 06m 04s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/dbc535839786447f858f2cc89b81441b : FAILURE in 1h 56m 28s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ec8694683a242c98eeee2d0c37de1ef : FAILURE in 1h 55m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4329f99f1b4c41848e4323b57df34ce8 : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":1},{"id":"29c6e3599676c49d78a5f99fadf5879ea23ae472","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-26 07:18:04.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f592ff3dd45d8fc1d963ebc7a2ec79cd86cd7f72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-26 09:31:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/db2480bff2584f208d2841778aaee11b : SUCCESS in 7m 50s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3c46c842815b4eff99c5fd0e69782ef4 : SUCCESS in 1h 04m 40s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/90e793551baf45e99d4c9341a64601d8 : FAILURE in 1h 28m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b264337e1504ac6bec5e0e9c8f5315a : SUCCESS in 1h 12m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28a2d5bc341140989c3a90e1c112b801 : SUCCESS in 1h 22m 45s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d160295cb7c4464886387c09b9f36e19 : FAILURE in 2h 06m 30s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf5106163835466db472d5792b7b3fa6 : SUCCESS in 1h 54m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff8dde55eb554f3cab67f9e38d3e9a9e : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":2},{"id":"7e3253029f8d149dbcb1a2911abb9bf4c6d1f356","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-05 08:34:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"31186948c6dbe3afab3c564573fe684def8e899c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 11:45:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b4010734266b4f9599f9951070502517 : SUCCESS in 8m 47s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ebbe9ef5fd1f4cc6944c9027f47461e1 : FAILURE in 2h 07m 15s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/dc9a3738d1fa42929a070a6f36e0fa4f : FAILURE in 1h 39m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ceebbe36f09543d58d7368a5b745f5a9 : SUCCESS in 1h 33m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/122f2e51508b47d8bc3209a81127ceab : SUCCESS in 1h 20m 11s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/649fd1283ed14cf2b5752f3531512f49 : FAILURE in 2h 03m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e477aefa6164d13884a6a5ff45f7e46 : TIMED_OUT in 3h 02m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df0f8f9824a74d60a0d108810ac39cc1 : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":2},{"id":"6b3598894a12546b5bf096d718a4bbc58c586512","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-06 08:53:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4bf6f6534b548bea069e235b31c4454587fdd0bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 11:14:10.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f47c5d2467cd4f7686f75ef7f98a0efb : SUCCESS in 9m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/90cca0a129254385980ec2a6c52888b5 : SUCCESS in 1h 37m 43s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d1d6783db8704982ab96c46c68d3e2fa : FAILURE in 1h 23m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcfd6cdae58244d8a2d28d7c07d5b562 : SUCCESS in 1h 31m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/564325e60dfa419584092ff1c263a2d9 : SUCCESS in 1h 16m 51s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/47ce07ef3c3e4c4e851e3b11a6934298 : FAILURE in 2h 13m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5682fcebf2074dbe9f97aee9a6647626 : SUCCESS in 1h 49m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af3c48c75ab84eb993077d181c72ba50 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":2},{"id":"eb4722f1f5b477c0a1c66452f0c2636761f825dd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-06 15:56:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1bc16fbea5582fe544dd9fe07b1976418f2c254e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 18:22:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e400211530e449eebaea891205443969 : SUCCESS in 9m 23s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/29b8c02db28c498dbfbd08053e682880 : SUCCESS in 1h 26m 20s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bd912f16220440a1a0334b2776324f8c : FAILURE in 1h 36m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f200f2b079b4b33a503d965bcdb37b8 : SUCCESS in 1h 24m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00b855bdb7b84ca8aa9cb3c81f27b494 : SUCCESS in 1h 27m 44s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d3f2912261b44dceb035dcb221c1f778 : FAILURE in 2h 09m 28s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aab254d83cab490c96bc56f5fcd81d38 : SUCCESS in 2h 20m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac1bacfa48b14661a301cd65d5e9a4b6 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"3d42a05ac8d0fd55501234f2595dff17fcf506bb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-08 15:50:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f642164e78a0a3e1d2d5199a4c52f06751be6365","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-08 18:23:55.000000000","message":"Patch Set 2:\n\ncentos things are failing here https://zuul.opendev.org/t/openstack/build/dc9a3738d1fa42929a070a6f36e0fa4f/log/logs/host/octavia-worker.service.journal-10-15-32.log.txt#3151","accounts_in_message":[],"_revision_number":2},{"id":"d4419bb2b1eb6ff60c496c4e0efdfa5048123ac5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 18:35:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c84d203aec4d429cab228821e313fa17 : SUCCESS in 10m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/83030ebc83b54990ba5034d21954e21b : SUCCESS in 1h 31m 23s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/650e7a995d0e4ef88ef84cf4eec1da75 : FAILURE in 1h 30m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e53710260c8c412ea2d413dcc996835c : SUCCESS in 1h 26m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54f15b6cc0e44788aa935c669fd8520d : SUCCESS in 1h 19m 17s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/403f42ba6cb84612bb76c8b2532d1b97 : FAILURE in 2h 33m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08026457a43344a7a9a57949bb4fdf3a : SUCCESS in 1h 56m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8eb15642c424d9187e7761aff43fe24 : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":2},{"id":"535a81b06ddc90a556b0db5bbb2171c3d963bd0d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-04 07:45:32.000000000","message":"Patch Set 2:\n\nlibvirt error on centos https://zuul.opendev.org/t/openstack/build/650e7a995d0e4ef88ef84cf4eec1da75/log/logs/host/nova-compute.service.journal-17-23-12.log.txt#6466","accounts_in_message":[],"_revision_number":2},{"id":"de38353ccf00df513b3b50240d23a005b5e742a0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-04 07:55:08.000000000","message":"Patch Set 2:\n\nvarious erros in the libvirt log https://zuul.opendev.org/t/openstack/build/650e7a995d0e4ef88ef84cf4eec1da75/log/logs/host/libvirt/libvirtd.log.txt","accounts_in_message":[],"_revision_number":2},{"id":"cda8e0d907c628ace94e7ab04a1d6f14d4dedca8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-01 17:51:50.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"92d1c49f65dc2d9c636931ff034c8f5750d55f89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 20:13:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8072d702c17a45088fb2b078334b9de0 : SUCCESS in 9m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/07bb4730dcd746138a7dd60c51ebaf38 : SUCCESS in 1h 15m 23s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c1df5f6f2fdf4e8894423bd0cee25941 : SUCCESS in 1h 21m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a2d1b1783c04ee181e33d82bb9c3ace : SUCCESS in 1h 19m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75a3ea54f05943b7af09550b7d1463b3 : SUCCESS in 1h 11m 34s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1ac623ad3edd4af39463955188be432a : FAILURE in 2h 10m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70d6a0003d70461192b08b0c0d207e08 : SUCCESS in 1h 57m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10063a8aedd342898aa19e8679c354a0 : SUCCESS in 10m 14s","accounts_in_message":[],"_revision_number":3},{"id":"cee6fd81fdcc00238631fc1ba8fab34b0a0ebf6b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-02 08:25:38.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"868abfe09e61f0851c4bf96fdf58121e6daeeb2d","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-03-02 08:30:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8ead5f5028bdbb15ed7faca66efcfeb36b6e4aff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 08:30:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d20e604d413f4efcdec9dd42c8114dd95f9f413b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 10:11:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a9ee2fa9ef4a42c4a06b98461890201b : SUCCESS in 9m 00s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9222d2b86e4044749ec36b58c3d7a871 : SUCCESS in 1h 03m 10s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8a473a3d38dd4749ba502fe67e6b0537 : SUCCESS in 1h 31m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/177255ff80e24181bcf9395d86bbcc67 : SUCCESS in 1h 15m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3f5d997165643c787dd4867b80f7a05 : SUCCESS in 57m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d5d8885084b499f8974ed967d3340d6 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":3},{"id":"3c8e2b00f2d67916842d8efa938ea77e11e3020b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 10:11:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"75616a153eaf9917422798de56beb9a8fe67f62f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 10:14:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2e2e747bbc744ac99f2c8fe81d9095e : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"78d204afb8760915bb32d15a318b20e02176473d","revisions":{"515504a28d42cae23cd4ce2a37c7d2623f1b7514":{"kind":"REWORK","_number":1,"created":"2020-12-25 16:09:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/14/768514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/14/768514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/14/768514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/14/768514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/14/768514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/14/768514/1"}}},"commit":{"parents":[{"commit":"3f7a572ebeae9ee76679c0cecb4ffaa0fa552fc3","subject":"Merge \"[doc] Adjust octavia docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3f7a572ebeae9ee76679c0cecb4ffaa0fa552fc3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-25 16:09:29.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-25 16:09:29.000000000","tz":120},"subject":"Use barbican for certificates storage","message":"Use barbican for certificates storage\n\nOctavia can do SSL termination only in case when barbican is available.\nWe should be able to add required configuration section only when barbican\nis also present in inventory\n\nChange-Id: Ie319fd02cdd60f8a8ac65f0508e9075f40839ae9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/515504a28d42cae23cd4ce2a37c7d2623f1b7514"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/515504a28d42cae23cd4ce2a37c7d2623f1b7514"}]},"branch":"refs/heads/master"},"b9a17365a97ce50936fb847d7b68611f5eeeb7f4":{"kind":"REWORK","_number":2,"created":"2020-12-26 07:18:04.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/14/768514/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/14/768514/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/14/768514/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/14/768514/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/14/768514/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/14/768514/2"}}},"commit":{"parents":[{"commit":"3f7a572ebeae9ee76679c0cecb4ffaa0fa552fc3","subject":"Merge \"[doc] Adjust octavia docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3f7a572ebeae9ee76679c0cecb4ffaa0fa552fc3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-25 16:09:29.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-26 07:17:55.000000000","tz":0},"subject":"Use barbican for certificates storage","message":"Use barbican for certificates storage\n\nOctavia can do SSL termination only in case when barbican is available.\nWe should be able to add required configuration section only when barbican\nis also present in inventory\n\nChange-Id: Ie319fd02cdd60f8a8ac65f0508e9075f40839ae9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b9a17365a97ce50936fb847d7b68611f5eeeb7f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b9a17365a97ce50936fb847d7b68611f5eeeb7f4"}]},"branch":"refs/heads/master"},"78d204afb8760915bb32d15a318b20e02176473d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-01 17:51:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/14/768514/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/14/768514/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/14/768514/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/14/768514/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/14/768514/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/14/768514/3"}}},"commit":{"parents":[{"commit":"44d0a6d39838cd76d55111096783bbcab838ec79","subject":"Add variable to override keypair setup python interpreter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/44d0a6d39838cd76d55111096783bbcab838ec79"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-25 16:09:29.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-01 17:51:50.000000000","tz":0},"subject":"Use barbican for certificates storage","message":"Use barbican for certificates storage\n\nOctavia can do SSL termination only in case when barbican is available.\nWe should be able to add required configuration section only when barbican\nis also present in inventory\n\nChange-Id: Ie319fd02cdd60f8a8ac65f0508e9075f40839ae9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/78d204afb8760915bb32d15a318b20e02176473d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/78d204afb8760915bb32d15a318b20e02176473d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
