)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~768873","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I2ecca967d78924ecd1ac03984e26ec87d1e6e4bf","project":"openstack/openstack-ansible-os_octavia","branch":"master","hashtags":[],"change_id":"I2ecca967d78924ecd1ac03984e26ec87d1e6e4bf","subject":"Add Victoria release info","status":"ABANDONED","created":"2021-01-02 08:30:43.000000000","updated":"2021-01-02 09:50:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"74e2671fc451eb486c163d69b1face0b73812520","_number":768873,"virtual_id_number":768873,"owner":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-01-02 08:40:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-02 08:40:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"73cd923d333b598d2053ae439883ea9712ee22b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2021-01-02 08:30:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b75689fd1342849c58034b3b0dbb9cb5ea125cbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-02 08:40:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fb64c47b4d8d4de4bdc1311ed4f9f74e : SUCCESS in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e578b1022bd4ae290011017283a1ecd : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23d67db052cd49f299b558d5a838ef70 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"74e2671fc451eb486c163d69b1face0b73812520","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2021-01-02 09:50:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40a9ad6ed60e5500c8f7c518c131d4e39eff326c","revisions":{"40a9ad6ed60e5500c8f7c518c131d4e39eff326c":{"kind":"REWORK","_number":1,"created":"2021-01-02 08:30:43.000000000","uploader":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"ref":"refs/changes/73/768873/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/73/768873/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/73/768873/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/73/768873/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/73/768873/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/73/768873/1"}}},"commit":{"parents":[{"commit":"3f7a572ebeae9ee76679c0cecb4ffaa0fa552fc3","subject":"Merge \"[doc] Adjust octavia docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3f7a572ebeae9ee76679c0cecb4ffaa0fa552fc3"}]}],"author":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-01-02 08:29:54.000000000","tz":480},"committer":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-01-02 08:30:36.000000000","tz":480},"subject":"Add Victoria release info","message":"Add Victoria release info\n\nAdd the lack of release information for Victoria, this patch added it.\n\nChange-Id: I2ecca967d78924ecd1ac03984e26ec87d1e6e4bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/40a9ad6ed60e5500c8f7c518c131d4e39eff326c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/40a9ad6ed60e5500c8f7c518c131d4e39eff326c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
