)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~772559","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~Ia2056cf287b666d4d3d8d36f06772c5117ca6bf7","project":"openstack/openstack-ansible-os_octavia","branch":"master","hashtags":[],"change_id":"Ia2056cf287b666d4d3d8d36f06772c5117ca6bf7","subject":"Add variable to override keypair setup python interpreter","status":"MERGED","created":"2021-01-26 15:23:03.000000000","updated":"2021-02-27 18:52:10.000000000","submitted":"2021-02-27 18:51:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772559","meta_rev_id":"7b9d5f7e08ea39f274b8c7f3ff305013602468dd","_number":772559,"virtual_id_number":772559,"owner":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-27 18:51:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-02-27 17:28:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-26 11:23:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2021-02-27 17:28:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-26 17:47:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-04 10:46:17.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2021-02-23 12:44:21.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-02-26 11:23:34.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"38a32f633b5acb81dab6341f6b1830a656833b0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-26 15:23:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8ad3c198b771031b2edc461841e5595305cb5de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 17:47:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5bc8205facbb4e0f9e4c7b75fcb8b5e6 : SUCCESS in 11m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0e4da77a04ad488d9b93a9d3f6bbe4c0 : FAILURE in 1h 27m 56s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d0b02778b25148ff8345a60cdbc5c0ef : FAILURE in 1h 46m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b91da8c919748159d02371c064b0dee : SUCCESS in 1h 30m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae731e13995e47af919b4d6089d50d3f : SUCCESS in 1h 07m 45s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/07961305490247778df431ad230605bf : FAILURE in 2h 13m 58s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26fef0037e8a4566be08558c138d1e70 : SUCCESS in 1h 43m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd63ccab6d2e448ba1612ca94573880f : SUCCESS in 12m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2f2e973e7f350221f5af8fd5c4b3a662d0524823","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-27 08:21:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bacda2d7a507326726598ebd017fbe48514f8b11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 10:38:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a5b73b618038485d90862aa4c4ce5f74 : SUCCESS in 7m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b016d0c7bf3a4e2eaf6a488076be8199 : FAILURE in 1h 21m 13s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/05d9e589a0244ac3ae436427be186863 : FAILURE in 1h 35m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d030da8bda3419b94eb5c2307494415 : SUCCESS in 1h 29m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67c9bd4c8e5447c8aa1cb8ccee7ea810 : SUCCESS in 1h 15m 48s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/07c8527259214a6a94973f56dd897272 : FAILURE in 2h 08m 46s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2213e507a9b43028cff36b4fb7bd76d : SUCCESS in 1h 54m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/224c736849f441be8915174b6696f8f8 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"fd5aeef2be7b114ed37c58da8b2af9e2be041160","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-04 10:46:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e591323aa5d3716ce717a09a68064b8ebf641a67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 13:06:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7a25ed19839f4a99bd0170a17854c835 : SUCCESS in 7m 29s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3d24bebc43714e27a02f361e066a439b : FAILURE in 1h 12m 43s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ae8eddd417de4c769bc5aa46d83d6fef : FAILURE in 1h 18m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/896df95c370045699a4562113003a898 : SUCCESS in 1h 22m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96c07108e7b64fc4bcfd620994eba4a7 : FAILURE in 33m 21s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/97292c69bab448778e52aa2830f1c111 : FAILURE in 2h 13m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8380e3ba9ad74c4884f86460c5f19228 : SUCCESS in 1h 56m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a8288d5283b46858d862ecf67c8c8b2 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"be9cdf37dd6dab36f89f4b029e94b65cf3e72446","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-02-23 08:16:20.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0c3877258128c95ed35dfe3de80a9cf6ff5c6f2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 10:35:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/17fe0cb48bc3499aa4d3cdbe530e55e4 : SUCCESS in 12m 34s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ed94a7df53b84e00b8016821ea3cb6be : RETRY_LIMIT in 25m 14s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4e76c335205e4d58bbb7c40042912361 : RETRY_LIMIT in 14m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48447cafec68420faab10393d017055b : RETRY_LIMIT in 15m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7eb8bc51cd741d09f9d4f12446e8aac : RETRY_LIMIT in 12m 12s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/79b4f9c617ac49189fff697ca5ab8484 : FAILURE in 2h 06m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23ae41a146154a1db2ebd6b6dec11407 : FAILURE in 1h 44m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0489739855c74834b40bdf0a23de2e3f : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":2},{"id":"1e89a0031dcaf402869686efd63ddbc26494655b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-02-23 11:15:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ea738f69eaa1240b19b44319314f3dc19bfd1de3","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-02-23 11:20:09.000000000","message":"Patch Set 3:\n\nInterested in opinions on changing \u0027openstack_keypair_setup_host\u0027 to pull from \u0027openstack_service_setup_host\u0027. Otherwise the former will need adding as a new var in the openstack-ansible repo.","accounts_in_message":[],"_revision_number":3},{"id":"d3de5a8d903aad994cf98f13491f65013191e015","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-23 12:44:21.000000000","message":"Patch Set 3: Code-Review+2\n\nWell, I\u0027d left openstack_keypair_setup_host. I think we might need more keypair generations like just in octavia. We may default it to openstack_service_setup_host somewhere (maybe in the integrated repo) but I see nothing bad in adding new variable.\n\nHowever, generating keypair like we do today requires exact same collections and set of modules like service setup, so having them combined might be ok as well. We may leave it like this right now and replace later if we feel it\u0027s needed.","accounts_in_message":[],"_revision_number":3},{"id":"a304ade96bccae1b99ee0680fedecea287f8ee63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 13:28:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fd130df4e38345e5b4136f9cfefde79a : SUCCESS in 8m 46s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d8ebd1b5be394695b6bd93c5c7df399e : FAILURE in 1h 13m 48s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b532ba637f8f4299a9e29b85daeb5275 : FAILURE in 1h 24m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5311991a52914220b4b68397f509c3e6 : SUCCESS in 1h 30m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b01d14efc9e041ad8bf8d45ea583d31b : SUCCESS in 1h 18m 22s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e5a5b2f446ba45a1850a2e71010f9723 : FAILURE in 2h 07m 21s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a9fab1204d643ebbbf04998a06303f2 : SUCCESS in 1h 48m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38f1ad43cbbc4fc18dc4ddbfd091901b : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":3},{"id":"d8f5ba2413f68cc3a8563f2b73c2f466060ce410","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-02-23 13:30:59.000000000","message":"Patch Set 3:\n\nThe remaining issues feel like https://bugzilla.redhat.com/show_bug.cgi?id\u003d1832710 on Debian, but unclear why this also impacts CentOS","accounts_in_message":[],"_revision_number":3},{"id":"9ca9a0be59ea4f2990bf920f9578092d796d75f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-02-26 08:24:58.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"7e25d079c20f144f1aa6f7f0149fbb9d04f9107f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 10:51:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6b5e20ac7f224d2286808e3215d8c809 : SUCCESS in 10m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/82ddbd2ee1c54da280f8f2d46fb5c52c : SUCCESS in 1h 21m 45s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1fbebbef51314fc0900fd984aa3333d9 : SUCCESS in 1h 41m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e8abf5a6a72467b8fd723eda437eece : SUCCESS in 1h 39m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/988e140cde8c4355ad1d3ffd9672508c : SUCCESS in 1h 32m 55s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a112d621c00349d9ae5a5d12c70fccf7 : FAILURE in 2h 20m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/078a286686874748a6739430e1b8ab58 : SUCCESS in 2h 05m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dba5eff0b9c46c2826b46fb79b53d06 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":4},{"id":"e7f972376c6e2d0f389e63a30ef9776a9c11591e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-26 11:23:34.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"baa4fa1ecc1ebdcacfc37dbf7517ab8d78b7ee8a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-27 17:28:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1b301a0b49a78d800bcd847b6c4cde66598fb931","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 17:28:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fa4156365604fee01a40cb3fc6834e360d2489d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 18:51:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/254bd76383f74dd0bdf180f8cb6a23f3 : SUCCESS in 9m 53s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1f7177c535ab41b5884f0bb7c329fca5 : SUCCESS in 1h 11m 52s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f59337cce63f4917a170407f6390b834 : SUCCESS in 1h 14m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91de24b66b694fada87bca1da63714fc : SUCCESS in 1h 19m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cc8ec5a77804c28836412ecd950c8ba : SUCCESS in 1h 09m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/312303eddf8c44e7917fbeeaada61c1e : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":4},{"id":"09dcc7480b2f95d043ae0b03faa5d62e620f9373","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 18:51:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"7b9d5f7e08ea39f274b8c7f3ff305013602468dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 18:52:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ec238e21d6f4633b347fb077b988c2c : SUCCESS in 49s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"44d0a6d39838cd76d55111096783bbcab838ec79","revisions":{"16f8f50254476d5dc3e1ebe8a30c8ed7bdc3df33":{"kind":"REWORK","_number":1,"created":"2021-01-26 15:23:03.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/59/772559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/59/772559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/59/772559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/59/772559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/59/772559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/59/772559/1"}}},"commit":{"parents":[{"commit":"3f7a572ebeae9ee76679c0cecb4ffaa0fa552fc3","subject":"Merge \"[doc] Adjust octavia docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3f7a572ebeae9ee76679c0cecb4ffaa0fa552fc3"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-26 15:09:06.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-26 15:23:01.000000000","tz":0},"subject":"Add variable to override keypair setup python interpreter","message":"Add variable to override keypair setup python interpreter\n\nWhen using a non-standard host for keypair setup (such as a utility\ncontainer) it is necessary to set a custom python interpreter\nwhich has access to openstacksdk.\n\nThis commit provides a variable to do this in the same style as\nused for service setup hosts.\n\nChange-Id: Ia2056cf287b666d4d3d8d36f06772c5117ca6bf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/16f8f50254476d5dc3e1ebe8a30c8ed7bdc3df33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/16f8f50254476d5dc3e1ebe8a30c8ed7bdc3df33"}]},"branch":"refs/heads/master"},"35a336f5002f4d5838a0c55257fdf54a75bf7db8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-23 08:16:20.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/59/772559/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/59/772559/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/59/772559/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/59/772559/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/59/772559/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/59/772559/2"}}},"commit":{"parents":[{"commit":"3f7a572ebeae9ee76679c0cecb4ffaa0fa552fc3","subject":"Merge \"[doc] Adjust octavia docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3f7a572ebeae9ee76679c0cecb4ffaa0fa552fc3"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-26 15:09:06.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-02-23 08:16:20.000000000","tz":0},"subject":"Add variable to override keypair setup python interpreter","message":"Add variable to override keypair setup python interpreter\n\nWhen using a non-standard host for keypair setup (such as a utility\ncontainer) it is necessary to set a custom python interpreter\nwhich has access to openstacksdk.\n\nThis commit provides a variable to do this in the same style as\nused for service setup hosts.\n\nDepends-On: https://review.opendev.org/777056\nChange-Id: Ia2056cf287b666d4d3d8d36f06772c5117ca6bf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/35a336f5002f4d5838a0c55257fdf54a75bf7db8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/35a336f5002f4d5838a0c55257fdf54a75bf7db8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ea0aaee6d31e47d768fd16935781285a174fc8ae":{"kind":"REWORK","_number":3,"created":"2021-02-23 11:15:06.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/59/772559/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/59/772559/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/59/772559/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/59/772559/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/59/772559/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/59/772559/3"}}},"commit":{"parents":[{"commit":"3f7a572ebeae9ee76679c0cecb4ffaa0fa552fc3","subject":"Merge \"[doc] Adjust octavia docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3f7a572ebeae9ee76679c0cecb4ffaa0fa552fc3"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-26 15:09:06.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-02-23 11:14:33.000000000","tz":0},"subject":"Add variable to override keypair setup python interpreter","message":"Add variable to override keypair setup python interpreter\n\nWhen using a non-standard host for keypair setup (such as a utility\ncontainer) it is necessary to set a custom python interpreter\nwhich has access to openstacksdk.\n\nThis commit provides a variable to do this in the same style as\nused for service setup hosts.\n\nChange-Id: Ia2056cf287b666d4d3d8d36f06772c5117ca6bf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ea0aaee6d31e47d768fd16935781285a174fc8ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ea0aaee6d31e47d768fd16935781285a174fc8ae"}]},"branch":"refs/heads/master"},"44d0a6d39838cd76d55111096783bbcab838ec79":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-02-26 08:24:58.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/59/772559/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/59/772559/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/59/772559/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/59/772559/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/59/772559/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/59/772559/4"}}},"commit":{"parents":[{"commit":"3f7a572ebeae9ee76679c0cecb4ffaa0fa552fc3","subject":"Merge \"[doc] Adjust octavia docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3f7a572ebeae9ee76679c0cecb4ffaa0fa552fc3"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-26 15:09:06.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-02-26 08:24:58.000000000","tz":0},"subject":"Add variable to override keypair setup python interpreter","message":"Add variable to override keypair setup python interpreter\n\nWhen using a non-standard host for keypair setup (such as a utility\ncontainer) it is necessary to set a custom python interpreter\nwhich has access to openstacksdk.\n\nThis commit provides a variable to do this in the same style as\nused for service setup hosts.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/777601\nChange-Id: Ia2056cf287b666d4d3d8d36f06772c5117ca6bf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/44d0a6d39838cd76d55111096783bbcab838ec79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/44d0a6d39838cd76d55111096783bbcab838ec79"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
