)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~779069","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I947c4b38810b8ad4ed22a5c6fc39ef9c5978a114","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"integrated_octavia","hashtags":[],"change_id":"I947c4b38810b8ad4ed22a5c6fc39ef9c5978a114","subject":"[WIP] Add integrated test for Octavia","status":"ABANDONED","created":"2021-03-06 03:36:36.000000000","updated":"2021-03-06 12:32:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8f7b0b05d6a1be1131de984efbb43209aa11afde","_number":779069,"virtual_id_number":779069,"owner":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-06 05:52:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-06 05:52:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-06 09:04:19.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"}],"messages":[{"id":"9f97436b0759d1b6709e6772ad31d48462d1a917","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-03-06 03:36:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d61efb27e62d1e0d02cabab9b3fdd87ffe011c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 05:52:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/788f809dbe2248e286f037f4c9e46f22 : SUCCESS in 9m 08s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6e1c437e8d014ce7b66c2027a6e0ffe4 : SUCCESS in 1h 06m 19s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e2e9ea9ca97d4f6c94965005c3af84cf : SUCCESS in 1h 11m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc31624be0404bc7af476e633d038d22 : SUCCESS in 1h 15m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06207ce957ea4be581af43477a20d721 : SUCCESS in 1h 06m 22s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/dadeaa1c56514df69b340abc12e8eba0 : SUCCESS in 2h 07m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aba07b4b1ff14980986672444bac69bc : SUCCESS in 1h 51m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29c61dabcc6b4941a5bdb88adcc1ddaa : SUCCESS in 7m 31s\n- openstack-ansible-deploy-aio_octavia_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8a09f02d204493d8ab2716a8956c408 : SUCCESS in 1h 26m 59s\n- openstack-ansible-deploy-aio_octavia_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a54a92b754054a62a252ad820d6d8311 : SUCCESS in 50m 54s","accounts_in_message":[],"_revision_number":1},{"id":"774b654df3ed3f6d5cad2ba3bd12951dfef5e76f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-06 09:04:19.000000000","message":"Patch Set 1:\n\nthia should be done automagically in zuul jobs for all os_\u003cservice\u003e repos with this https://github.com/openstack/openstack-ansible/blob/master/zuul.d/playbooks/pre-gate-scenario.yml#L39-L44\n\nCheck in the zuul log when SCENARIO gets set to.","accounts_in_message":[],"_revision_number":1},{"id":"7379822559bd1bd5daecd74375a97d69fd625b79","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-06 09:08:35.000000000","message":"Patch Set 1:\n\nexample here from your other patch 779068 \n\nhttps://zuul.opendev.org/t/openstack/build/8d8ae706220345a39130c49e3b8815d5/log/job-output.txt#2833-2852","accounts_in_message":[],"_revision_number":1},{"id":"1fdf48cd8d69c2f8aa4d6fc825ca8f3765956c90","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-03-06 12:31:59.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e example here from your other patch 779068 \n\u003e \n\u003e https://zuul.opendev.org/t/openstack/build/8d8ae706220345a39130c49e3b8815d5/log/job-output.txt#2833-2852\n\nI was *hoping* something like that was taking place since I saw overrides, just wasn\u0027t sure and hadn\u0027t looked at existing logs. I was focusing on the job name. Thanks for the tip!","accounts_in_message":[],"_revision_number":1},{"id":"8f7b0b05d6a1be1131de984efbb43209aa11afde","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-03-06 12:32:11.000000000","message":"Abandoned\n\nThis is already being done!","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14ea9bd2dc26dafdf0ac0f60dbac7c4057281478","revisions":{"14ea9bd2dc26dafdf0ac0f60dbac7c4057281478":{"kind":"REWORK","_number":1,"created":"2021-03-06 03:36:36.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/69/779069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/69/779069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/69/779069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/69/779069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/69/779069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/69/779069/1"}}},"commit":{"parents":[{"commit":"e844e8adfcb0871966ec019979298c21f7da2699","subject":"Merge \"Move octaiva pip packages from constraints to requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/e844e8adfcb0871966ec019979298c21f7da2699"}]}],"author":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2021-03-06 03:35:01.000000000","tz":0},"committer":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2021-03-06 03:36:35.000000000","tz":0},"subject":"[WIP] Add integrated test for Octavia","message":"[WIP] Add integrated test for Octavia\n\nThis patch set aims to kick the tires on an integrated test\nwith Octavia deployed. The goal is to test amphora deployment\nand connectivity (if possible). Do not merge!\n\nChange-Id: I947c4b38810b8ad4ed22a5c6fc39ef9c5978a114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/14ea9bd2dc26dafdf0ac0f60dbac7c4057281478"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/14ea9bd2dc26dafdf0ac0f60dbac7c4057281478"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
