)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~784614","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~stable%2Fvictoria~Ibd5f3d2ca25f9bb880b0c535c59ef430bd1043be","project":"openstack/openstack-ansible-os_octavia","branch":"stable/victoria","hashtags":[],"change_id":"Ibd5f3d2ca25f9bb880b0c535c59ef430bd1043be","subject":"Omit amp_ssh_access_allowed and remove amp_image_id options.","status":"MERGED","created":"2021-04-02 20:25:28.000000000","updated":"2021-04-09 19:22:49.000000000","submitted":"2021-04-09 19:21:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784614","meta_rev_id":"9ccc732707acc20f580a533d0d8dcc69622483d4","_number":784614,"virtual_id_number":784614,"owner":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-09 19:21:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-04-03 06:13:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-06 13:42:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-06 13:42:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-02 20:25:28.000000000","updated_by":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"reviewer":{"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"},"state":"REVIEWER"},{"updated":"2021-04-02 22:30:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-03 06:13:10.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-04-06 13:42:55.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"07474a918f438d7934affbef3e1ce34d44a0b191","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2021-04-02 20:25:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d07a6b20b7ad6138edc2ce1b2a63ab57c2e5e910","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 22:30:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/603f32dbd17844088d3ef9806f4ce351 : SUCCESS in 9m 41s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b783b8898ca74c5c8f15e85fe681b362 : SUCCESS in 1h 24m 46s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5ec864b456da4a05a2dc13a90f43c720 : SUCCESS in 1h 33m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76b1469e0c4d443186315b3232775627 : SUCCESS in 1h 27m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/537672d4a73543c6b467d92d22d0f7ee : SUCCESS in 1h 20m 44s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e5f4bfaf7dc14ea48459a781fe51f6c4 : SUCCESS in 1h 59m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6c7703f441740c4982fba63de0f7f17 : SUCCESS in 1h 59m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce8b3a54a05947a1956145e2715bb01e : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61f1e85279d24058893b2c85593aa6b2 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"98ac9d4110625132acd7f6e88819d750531dad6e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-03 06:13:10.000000000","message":"Patch Set 1: Code-Review+2\n\nWhile option is really deprecated, not sure how good is that to backport variable removal. considering we have release note - I think that\u0027s fine...","accounts_in_message":[],"_revision_number":1},{"id":"91919b6e99d70ad4cd26d12385eaeda04511db80","author":{"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"},"date":"2021-04-06 05:44:57.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review+2\n\u003e \n\u003e While option is really deprecated, not sure how good is that to backport variable removal. considering we have release note - I think that\u0027s fine...\n\nMy patch was created in December. It is just laying around for 4 month as we have since then not been able to get test not failing.","accounts_in_message":[],"_revision_number":1},{"id":"f34635428b1e5718ce18d9c2f93bb26facdded60","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-06 13:42:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ee71e5754dc6f75746c5499e9e4d09030b7e497","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 13:43:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7221fa751d1e5ef3336163e2444d4a155a1ee7d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 15:34:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/db1e80eed2a84f38912391345b868124 : SUCCESS in 13m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/83bd33130c774623bb0e2cbc8c85b206 : SUCCESS in 1h 20m 55s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5db7a3020e44486b918a2095e6f935e4 : POST_FAILURE in 1h 35m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a69638ff1f24fc7a2479771f268f9e4 : SUCCESS in 1h 30m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7500bfc02f2340b796b3c1ded0d8b5de : SUCCESS in 1h 18m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/797751cb6a224dbeb14666ac240cf373 : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0014c6a117f84a4890ba3fd4c5f8bc70 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"aa47393eb17a913389e25345b439ad02a9a901ad","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-06 16:09:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4f306eb4a1e88132d4ce7eebf4210ae2382d3dca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 18:13:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bb3d896bda83434789a10ff532dde511 : SUCCESS in 13m 01s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d4daf8d1779c4829877f82cb721bf565 : SUCCESS in 1h 18m 25s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a5837ef3fd7a4612bfc82a76355a7dd4 : FAILURE in 1h 27m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a801877fc1cc4250957efaa877f1abd5 : SUCCESS in 1h 25m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff42e84f5cf94efbb99f092710da454a : SUCCESS in 1h 17m 06s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/77c24c704bc54f62ab35021218f1d603 : SUCCESS in 1h 55m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a219e4cd8a74fd6a0211a3659eafda4 : SUCCESS in 1h 47m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/188f6ba1308f44449e113ad08ed4a4b9 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5da9c14abcd74efba3c9c5dd2498e03f : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5ff2ff6259eb3094bd8e3fc2d500f070f8cccd6e","author":{"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"},"date":"2021-04-07 09:07:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"76ee5fe3c66f7592e0359dfb0f42cda7f23859a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 11:27:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7b447b51232a4e4cadee5d7ad87f3885 : SUCCESS in 9m 33s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/61cc709cacd748c191ff6c2394242ac4 : SUCCESS in 1h 12m 58s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/17075005c2764b419310e093b86a00ee : FAILURE in 1h 24m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/621e93dd31524093a2532d9c7ce8b422 : FAILURE in 22m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58388a584f124ebdb6ecd6d46867d453 : SUCCESS in 1h 09m 55s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7f8f3e8e0c394c97982abd5706e542b4 : SUCCESS in 2h 15m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85f6fcfa566d4be58a7ad9d2fb783d99 : FAILURE in 14m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1064c20e8c8445c595d68aec65a0401c : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3906a22d8d1c492eb3e3a1a2d72014ff : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"742491243b8dd3f2e9f8e6111db20ea228e15da7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-09 06:03:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fda1dff86e9403854e82a3ae0123557de47a7d19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 08:38:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5c569cbb5e3b4f38b215c12fdfbec464 : SUCCESS in 11m 58s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/fc075e5450784dafa90f916d1bc7c8b4 : SUCCESS in 1h 11m 35s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/170d96a3718f4465b8e1365e3f20c08a : SUCCESS in 1h 11m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a4b481252f243cd891192f012d40136 : SUCCESS in 1h 30m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ec199c84c3d4ba586819acf2d2c2f6f : SUCCESS in 1h 18m 23s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e252f1be8980447d99d946b803fbbd10 : SUCCESS in 2h 16m 02s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05c96a3d5fda41a2a28f4e765442d432 : SUCCESS in 1h 50m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e367011780d64959b4fbb86859559ba1 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a770eb3a092f417798860e55f43267a6 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"6f23c7cd3da4e72e5e3787233ac28d1573145a23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 08:39:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23acf30c36158e7a5bf0e4e543b4ac2b38ab99fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 10:18:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/509b663d8eda499b951f45c75aec4ada : SUCCESS in 11m 33s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1aad6ce16a9d4f0a8e03c85b4ae8fed7 : FAILURE in 1h 16m 36s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/40c89bb360ec4a68919a7508a2b8bce3 : SUCCESS in 1h 31m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/990086d46f2d4467ad40401852712860 : SUCCESS in 1h 23m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf46bdaa6fd5419496d4c72a01a948c9 : SUCCESS in 1h 18m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20f51f2ac1e247328d3eb6c00c0157f3 : SUCCESS in 9m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a856e5757dcb426dae66cd35cfe2d4e5 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d57baea057e8dae8ea01bd230512721bf5db88f3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-09 14:18:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"733d66bb6c217c27f62b477369fd60f4780b1d1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 17:46:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/882b69d2e00d42ce95b65d7c05471527 : SUCCESS in 11m 29s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0df59f4ed7e448249acff84701121eda : SUCCESS in 1h 25m 41s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7db8a98dd533489a9e9f0f8437e7c5cb : SUCCESS in 1h 25m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15494bc9394845eea86f760c00faac1b : SUCCESS in 1h 22m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a27e7427fb7e46f5a6bdbdf7ebbb732e : SUCCESS in 1h 30m 51s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7e25bfece54449cbaf864e3afd19cdcf : SUCCESS in 2h 11m 45s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd048b409a1544b49f7afd0146c44a77 : SUCCESS in 1h 49m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be501cf9e68b49d1859aa027298dacd0 : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3699c61bf46e4cf1bc6eb6ac2ca30711 : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":1},{"id":"9d6d71f411f10eb369507c81056a5eb4031dbe74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 17:46:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"838a9ca420610601c311317195413e57cd23d94c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 19:21:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e5e16f9ca03b470fb81849db9a4ec936 : SUCCESS in 10m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/54010ed784ff4b758885108f06d7916c : SUCCESS in 1h 19m 51s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ea054ddd820c4d9e8c9b701042d9a015 : SUCCESS in 1h 28m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a70977fb731b40a88bae8b6f32d22754 : SUCCESS in 1h 26m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8a98a881bb14aed8461ca5ce429701d : SUCCESS in 1h 14m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c9430f303754feabf6c3c8b082ad9ff : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/901aebfccaa94b3dbf981dd877cfc8bb : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"fff07039ccf107e93e46618d2f40dcba8b46a31d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 19:21:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9ccc732707acc20f580a533d0d8dcc69622483d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 19:22:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43ae345be3c1441888789b4689f27780 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/253f7439e8b84fb3826f0b217014771f : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6fb63e341f99eb022e9d770dc934a964b538308f","revisions":{"6fb63e341f99eb022e9d770dc934a964b538308f":{"kind":"REWORK","_number":1,"created":"2021-04-02 20:25:28.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/14/784614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/14/784614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/14/784614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/14/784614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/14/784614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/14/784614/1"}}},"commit":{"parents":[{"commit":"3d0ca7da881c5452a6438ef267f3f104987863f2","subject":"Use barbican for certificates storage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/3d0ca7da881c5452a6438ef267f3f104987863f2"}]}],"author":{"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","date":"2020-12-17 10:47:36.000000000","tz":60},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2021-04-02 20:25:07.000000000","tz":-300},"subject":"Omit amp_ssh_access_allowed and remove amp_image_id options.","message":"Omit amp_ssh_access_allowed and remove amp_image_id options.\n\nOctavia complains about option amp_ssh_access_allowed to be deprecated. See\nhttps://docs.openstack.org/octavia/ussuri/configuration/configref.html#controller_worker.amp_ssh_access_allowed\nThe octavia_ssh_enabled OSA variable is instead used to either write the\namp_ssh_key_name configuration option or not.\n\nThe configuration option amp_image_id in Octavia is deprected and image tags\nshould be used instead. Therefore octavia_amp_image_id is removed.\n\nChange-Id: Ibd5f3d2ca25f9bb880b0c535c59ef430bd1043be\n(cherry picked from commit 50b83c79274d5cba9869ca4e52ac96a731445a84)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/6fb63e341f99eb022e9d770dc934a964b538308f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/6fb63e341f99eb022e9d770dc934a964b538308f"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
