)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~787336","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~Ib5627dc3b37626e056c3cfe9ce54ee6a7ff25dd5","project":"openstack/openstack-ansible-os_octavia","branch":"master","topic":"787336","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-05-11 13:52:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib5627dc3b37626e056c3cfe9ce54ee6a7ff25dd5","subject":"Make octavia_provider_network better configurable","status":"MERGED","created":"2021-04-21 09:41:20.000000000","updated":"2022-05-11 13:54:16.000000000","submitted":"2022-05-11 13:52:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787336-787336","meta_rev_id":"e7934342ccc32b5d69c19babc9966305a502757c","_number":787336,"virtual_id_number":787336,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-11 13:52:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2022-05-11 10:37:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-11 08:24:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2022-05-11 12:06:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-21 11:26:22.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2021-04-21 12:53:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-11 08:24:00.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"9f42574d15c766fd2788218c9830eb0aa6c36abf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-21 09:41:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3c9118b2e75199dcbebc0e497b0639c0d9d1a58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-21 10:27:13.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0f39e45e0464b334d36c8d0b482da6691c653516","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-04-21 11:26:22.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7a861158258e5d4569cdef9567246eefe9f18da8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 12:53:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/30add84449ca425c8c248575dca8cf14 : SUCCESS in 11m 12s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/103b1d26d4ca4cdd8a21d9dc933edc22 : FAILURE in 1h 00m 54s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9dfa48fc570a481f82db3e4ecf98ef09 : FAILURE in 1h 18m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/945370544fff4d0e8d9db2373ea17f11 : FAILURE in 1h 12m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd84bc39ed23466b916661d8c9caa7a7 : FAILURE in 1h 08m 02s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9254aadfeab74bd7b2040fe5e7eecc74 : SUCCESS in 2h 06m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98a5448d7f2849279728a029cfded947 : SUCCESS in 2h 08m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6154be4bbcce4a189abb4b43762a8475 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":2},{"id":"dc4172ec0ef25d8738dc18a56f9dad4e9e77547f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-21 18:18:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e9911007480b770931aa9bfc454221e4d588d716","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-21 18:20:00.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"db1edb0e135ba0451bb76fa9ff2b4823400d1aed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 20:50:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1759a57600e34ca18b1609cd4efe3c95 : SUCCESS in 13m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/29ac8fc94d5347fbb6680d2412ae5b0b : SUCCESS in 1h 02m 08s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/919ccfec7b684fa88b51b5491e55d5e3 : SUCCESS in 1h 10m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df1eaf1e518444d18eae236f99ce3306 : SUCCESS in 1h 09m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43b9b97a3f3b491c8eef8c6a893a6466 : SUCCESS in 59m 54s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4ac0e3aef5d0454e9facfd2d1d19af28 : SUCCESS in 2h 11m 28s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3263b43ab7b94b248d2e27065084233e : SUCCESS in 1h 32m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1598b75bbda94fff905cbb790decf956 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":3},{"id":"3985f9325be1ce832af8f707d19058b8eb86e151","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-14 14:42:50.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f2b383d76d39eb6d626508308bff7a4b06ad2588","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 16:48:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b952da3d38c8469b9996f42421a7f029 : SUCCESS in 15m 34s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3054f2f10df148329ee6d142bcbcb9f4 : FAILURE in 1h 08m 31s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/96b164ae11324d698cc766f983d69428 : FAILURE in 48m 35s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8785eabdcd8546d79a89872b4a1c066b : FAILURE in 1h 17m 03s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/edcb8fa635ab4305ac5a7f6753b19cda : FAILURE in 1h 19m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/910962e199cb4bc29d2fa3471a7d4f29 : FAILURE in 1h 14m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e72c988d6524440b8453abcb3709086 : FAILURE in 1h 08m 21s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9856b5f744fc47ab9f6b36ce7c22fef7 : FAILURE in 2h 03m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfdf5e009a81433bafe013c831131f1d : FAILURE in 1h 48m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9edc554076d24c12a74806566664eb9c : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":4},{"id":"38f5163809fdc57b831efad615eb0263fe1f847c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-10 16:45:59.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b3a31acfcac26e1d98edb3f372a375b4f11f3fd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 19:04:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e6503a4e8904a478d62edc9db6ac83c : SUCCESS in 12m 46s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/bbeca0b8600d4c4c8ffe281accc26435 : SUCCESS in 1h 23m 44s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e70c7c94b3fa428490cdd0eb7ca08e50 : SUCCESS in 1h 28m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d493b78360e42dfa379bcf093afdd00 : SUCCESS in 1h 39m 32s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/4155a4d7056a416cb72e0ca2322c8427 : SUCCESS in 1h 36m 49s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/08c5f967908043d6b38e86e14db49924 : SUCCESS in 2h 07m 42s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d401008ba244c3da1b0644faacca17a : SUCCESS in 2h 10m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e2541945c2e411b91d29337bf73c9e1 : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":5},{"id":"d6ff8f19551f259514cbf75428f7397859492d34","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-11 08:24:00.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"3e566940b5c74f5ff8ffa13b04c14785866cb1bd","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-05-11 10:37:50.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"dec0937351ad02a2dec1058c4a3b2c2b0add4229","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-05-11 12:06:31.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"4cd05d02bf65d4f0dc9f1af11aa6c658cb2292be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 12:09:08.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c9ac174d911ecbce216b75d5d917826a9ebfe380","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 13:52:42.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b82f26278ec485e973f1ca1991323d9 : SUCCESS in 14m 05s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a00532f8e413400ebceeba4b723f57cb : SUCCESS in 1h 08m 15s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e164fe98a2404a18b4bb58d6a662a783 : SUCCESS in 1h 20m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a175f3b264df4402aad93e9c61ccfbde : SUCCESS in 1h 20m 21s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/f418815f9df34f48aae2a82607789f6d : SUCCESS in 1h 37m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/907d6e271d6244d9bb38b8356ab11d57 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":5},{"id":"5e48f10808de699066a42a0b924396281f4c179a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 13:52:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"e7934342ccc32b5d69c19babc9966305a502757c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 13:54:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f1959966443417c9172717a22fc452b : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"941e671e063099f151023896daf268d742a106f2","revisions":{"bd19eb693170a51bc01ee78e08f3841149f1b357":{"kind":"REWORK","_number":1,"created":"2021-04-21 09:41:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/36/787336/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/36/787336/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/36/787336/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/36/787336/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/36/787336/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/36/787336/1"}}},"commit":{"parents":[{"commit":"50b83c79274d5cba9869ca4e52ac96a731445a84","subject":"Omit amp_ssh_access_allowed and remove amp_image_id options.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/50b83c79274d5cba9869ca4e52ac96a731445a84"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-21 09:41:08.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-21 09:41:08.000000000","tz":180},"subject":"Make octavia_provider_network better configurable","message":"Make octavia_provider_network better configurable\n\nFor vlan scenraio we can\u0027t use  octavia_provider_network_name for\noctavia_provider_network but it\u0027s pretty big override, which might be\nmore handy with having an extra variable, that will be used inside it.\n\nChange-Id: Ib5627dc3b37626e056c3cfe9ce54ee6a7ff25dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/bd19eb693170a51bc01ee78e08f3841149f1b357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/bd19eb693170a51bc01ee78e08f3841149f1b357"}]},"branch":"refs/heads/master"},"8dc4e8c0e24d32dad513fadb7cba096240e01dc8":{"kind":"REWORK","_number":2,"created":"2021-04-21 10:27:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/36/787336/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/36/787336/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/36/787336/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/36/787336/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/36/787336/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/36/787336/2"}}},"commit":{"parents":[{"commit":"50b83c79274d5cba9869ca4e52ac96a731445a84","subject":"Omit amp_ssh_access_allowed and remove amp_image_id options.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/50b83c79274d5cba9869ca4e52ac96a731445a84"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-21 09:41:08.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-21 10:27:07.000000000","tz":0},"subject":"Make octavia_provider_network better configurable","message":"Make octavia_provider_network better configurable\n\nFor vlan scenraio we can\u0027t use  octavia_provider_network_name for\noctavia_provider_network but it\u0027s pretty big override, which might be\nmore handy with having an extra variable, that will be used inside it.\n\nChange-Id: Ib5627dc3b37626e056c3cfe9ce54ee6a7ff25dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8dc4e8c0e24d32dad513fadb7cba096240e01dc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/8dc4e8c0e24d32dad513fadb7cba096240e01dc8"}]},"branch":"refs/heads/master"},"aeac785164e2e8e2a3946b24515e4e02317a1274":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-04-21 18:20:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/36/787336/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/36/787336/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/36/787336/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/36/787336/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/36/787336/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/36/787336/3"}}},"commit":{"parents":[{"commit":"50b83c79274d5cba9869ca4e52ac96a731445a84","subject":"Omit amp_ssh_access_allowed and remove amp_image_id options.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/50b83c79274d5cba9869ca4e52ac96a731445a84"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-21 09:41:08.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-21 18:19:56.000000000","tz":0},"subject":"Make octavia_provider_network better configurable","message":"Make octavia_provider_network better configurable\n\nFor vlan scenraio we can\u0027t use  octavia_provider_network_name for\noctavia_provider_network but it\u0027s pretty big override, which might be\nmore handy with having an extra variable, that will be used inside it.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/787199\nChange-Id: Ib5627dc3b37626e056c3cfe9ce54ee6a7ff25dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/aeac785164e2e8e2a3946b24515e4e02317a1274"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/aeac785164e2e8e2a3946b24515e4e02317a1274"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6f61dd051cb99267e16e47ccdc8444149a0b2120":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-14 14:42:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/36/787336/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/36/787336/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/36/787336/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/36/787336/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/36/787336/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/36/787336/4"}}},"commit":{"parents":[{"commit":"b37ab7592e0d19b3c2f97e5cf5acc751313ce9bb","subject":"Changed minversion in tox to 3.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/b37ab7592e0d19b3c2f97e5cf5acc751313ce9bb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-21 09:41:08.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-07-14 14:42:50.000000000","tz":0},"subject":"Make octavia_provider_network better configurable","message":"Make octavia_provider_network better configurable\n\nFor vlan scenraio we can\u0027t use  octavia_provider_network_name for\noctavia_provider_network but it\u0027s pretty big override, which might be\nmore handy with having an extra variable, that will be used inside it.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/787199\nChange-Id: Ib5627dc3b37626e056c3cfe9ce54ee6a7ff25dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/6f61dd051cb99267e16e47ccdc8444149a0b2120"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/6f61dd051cb99267e16e47ccdc8444149a0b2120"}]},"branch":"refs/heads/master","description":"Rebase"},"941e671e063099f151023896daf268d742a106f2":{"kind":"REWORK","_number":5,"created":"2022-05-10 16:45:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/36/787336/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/36/787336/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/36/787336/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/36/787336/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/36/787336/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/36/787336/5"}}},"commit":{"parents":[{"commit":"4ba4409e463f96cd28b1ca915a25031d183863d9","subject":"Change octavia private key ciphers to type \u0027auto\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/4ba4409e463f96cd28b1ca915a25031d183863d9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-21 09:41:08.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-10 16:45:51.000000000","tz":120},"subject":"Make octavia_provider_network better configurable","message":"Make octavia_provider_network better configurable\n\nFor vlan scenraio we can\u0027t use  octavia_provider_network_name for\noctavia_provider_network but it\u0027s pretty big override, which might be\nmore handy with having an extra variable, that will be used inside it.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/787199\nChange-Id: Ib5627dc3b37626e056c3cfe9ce54ee6a7ff25dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/941e671e063099f151023896daf268d742a106f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/941e671e063099f151023896daf268d742a106f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
