)]}'
{"id":"openstack%2Fopenstack-ansible-os_octavia~787670","triplet_id":"openstack%2Fopenstack-ansible-os_octavia~master~I22c56f32d7308803e9363f9375d7f6206ccecd41","project":"openstack/openstack-ansible-os_octavia","branch":"master","hashtags":[],"change_id":"I22c56f32d7308803e9363f9375d7f6206ccecd41","subject":"Set octavia_amp_image_owner_id by default","status":"MERGED","created":"2021-04-23 07:13:57.000000000","updated":"2021-04-26 19:02:24.000000000","submitted":"2021-04-26 19:01:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787670","meta_rev_id":"0d17b6e39895686b86940897c927bb9407c0b7f7","_number":787670,"virtual_id_number":787670,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-26 19:01:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2021-04-26 17:26:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-26 16:32:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2021-04-26 17:26:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-23 09:24:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-26 16:32:39.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-04-26 17:26:58.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"c0187397dcca7f718739fbb67cf792e7a36f222d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-23 07:13:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c18596b0863a9f5703d53269aa2cb4b4079aa35e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 09:24:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bf0e6daf5e8d44169203b9323657a828 : SUCCESS in 9m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/50b679be89ba41cc9a52ae8efa15d44a : FAILURE in 1h 06m 12s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e37a3a636794480fb0fddf355c240cb1 : FAILURE in 1h 14m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a30e8069bf1c43ea893927a0e0f4c383 : FAILURE in 1h 13m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48b6b0b5c03242878c1e0f128b0cae52 : FAILURE in 49m 43s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bab39b13cac94abb97aad1ff1f50d1ab : FAILURE in 2h 02m 16s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4347c9aed99b4941aaf51c9c6ee5fbb9 : FAILURE in 1h 43m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b7f34660d2d4cb78d97e0b0cdcb7f86 : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":1},{"id":"04eb2e68b0a5d41736f774ed2d4d06dac31b9043","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-23 11:28:26.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7b67bac0839963dde0eddca14426a8e9854308a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 13:14:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0a54e8d307aa4ddcab3a32ea82d7682f : SUCCESS in 12m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cc0ff2dc5a7643d7873b7ff7c82fccb4 : FAILURE in 1h 12m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1fb700174f6e4c4aa0272f8a7d6e41a3 : FAILURE in 1h 02m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/296693a8a6ec48ca829a85bd6c2b0878 : FAILURE in 1h 09m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acb94f6b6bc14e96b6d1e75e1d3bfc64 : FAILURE in 56m 58s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3a9a8c53ee0041a9bd5535ce7c5626e2 : FAILURE in 1h 19m 26s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1200a423ce7488dbf7f826cceb158c4 : FAILURE in 1h 22m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9b1b67b31b84c8285a2fb86c0003fae : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":2},{"id":"18ef24ac02b2102122d72aa30941ceef91fff2e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-23 13:45:13.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"2c707e88c409b6b3d2e020df2daa204233e9534c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 15:55:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/27c15f3f3cbe4d6e987c4b0d1108462d : SUCCESS in 13m 13s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/58d2a795b6324c39b78d7508ed9b773e : SUCCESS in 1h 35m 06s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8fe647c4c57147b4b1566f5c94515fc7 : RETRY_LIMIT in 14m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3bc88f0860146b99476480250617b39 : SUCCESS in 1h 17m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe2e4a8d842e4aa4befcf25460f33058 : SUCCESS in 1h 22m 00s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e5e09898b2a5463eacfaca5f7d799ae9 : FAILURE in 19m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fb283aa29b641cfb042c791e0ec2e70 : SUCCESS in 2h 02m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae6daa9206924ed39067e8b395800cad : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":3},{"id":"4875b59844ab4ed3ea298d74f6ac615a6de4cdcf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-23 20:12:47.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9ed510ec23e85d7573598ba83b3f968e9a228fbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 22:21:03.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f18c192b2eae47a1b8479c77ec0ad704 : SUCCESS in 8m 29s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1fd109881ca64e21a4491df7438e1611 : FAILURE in 1h 15m 03s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c508b127d97d41a3a415505cc08c0365 : RETRY_LIMIT in 13m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/edf093e9f3ab48f396e094092f2df597 : SUCCESS in 1h 19m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18cd69ce87fd404cbbf8e2369ede56a1 : SUCCESS in 1h 20m 10s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a3f6a4573c544e50802785c4742d84e8 : FAILURE in 14m 32s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/defd51c82832462390d3fb734787881b : SUCCESS in 1h 57m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe3be556b38644a08d330205b9dcf09a : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":3},{"id":"046982d2668e3bdea3e2514864a4faa33be40832","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-26 07:29:12.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"35d753cd7fd4c4e564c2b65726f5b7ea876cd65b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 09:25:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c3158644bf254d56b8c0345163c1f37c : SUCCESS in 13m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e4f83dbef69a4077b8d4bedee0cd5726 : SUCCESS in 1h 11m 06s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/920378c61fad466d82b6d9e1472157b3 : SUCCESS in 1h 14m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4854d7abd5c34f0b806f0724d88df5d0 : SUCCESS in 1h 09m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1aab8fba80be48588c42b5ef46c03e71 : SUCCESS in 1h 07m 45s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e031c34b9bd74953adb74ed0b12b4b1f : SUCCESS in 1h 51m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7909434e5fed43e08051b5a85fc29a9c : SUCCESS in 1h 32m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6413983d58ef4228b58ec453b5dcf0af : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":3},{"id":"9dd4f2ec7adcb23cc2e6ecf9a0c7333c6f08170f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-26 16:32:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c6dc598ecac5e51da3019c62b4b6fc534ef831e6","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-04-26 17:26:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"84236c06bff809e89b1cff8a2723364b204b12a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 17:27:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e6cb6d082fca7f099855f615d9ac02c4500653b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 19:01:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0efba8af8fc84edd8191eeb8aabf4f5e : SUCCESS in 8m 43s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/687c2fc16a1a43d183cf067e97ba9910 : SUCCESS in 1h 05m 53s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/58d6b42100f34ce88fcdc4c7edddd596 : SUCCESS in 1h 09m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0259c71927de43ff932c00515df723ce : SUCCESS in 1h 27m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abf7e0163382418cbcd53b486a3045c3 : SUCCESS in 59m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b1ef292c3c471c92bd6a57530b4318 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":3},{"id":"6e067ddfa79ec7fa699e6f55748ab96bd7dcaf2f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 19:01:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0d17b6e39895686b86940897c927bb9407c0b7f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 19:02:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df74ef61f01d411182308d605bc1599a : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9ef492e3b2663d035061745b23b471f3c8d5da27","revisions":{"ec5c3b7a3c51129fb12395f698c862abed33e6f8":{"kind":"REWORK","_number":1,"created":"2021-04-23 07:13:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/787670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/70/787670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/70/787670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/70/787670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/70/787670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/70/787670/1"}}},"commit":{"parents":[{"commit":"370ea30b99e57c711ede72d90fd2323385d4a2b1","subject":"Merge \"Remove unused handlers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/370ea30b99e57c711ede72d90fd2323385d4a2b1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-23 07:13:41.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-23 07:13:41.000000000","tz":180},"subject":"Set octavia_amp_image_owner_id by default","message":"Set octavia_amp_image_owner_id by default\n\nThis patch sets the octavia_amp_image_owner_id default so that the\nOctavia configuration file has amp_image_owner_id populated.\nThat is important security setting for production deployments that helps\nto avoid faked images being used\n\nChange-Id: I22c56f32d7308803e9363f9375d7f6206ccecd41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ec5c3b7a3c51129fb12395f698c862abed33e6f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/ec5c3b7a3c51129fb12395f698c862abed33e6f8"}]},"branch":"refs/heads/master"},"788f51cac6f26227fdcdc262b963d96a16286e5e":{"kind":"REWORK","_number":2,"created":"2021-04-23 11:28:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/787670/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/70/787670/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/70/787670/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/70/787670/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/70/787670/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/70/787670/2"}}},"commit":{"parents":[{"commit":"370ea30b99e57c711ede72d90fd2323385d4a2b1","subject":"Merge \"Remove unused handlers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/370ea30b99e57c711ede72d90fd2323385d4a2b1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-23 07:13:41.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-23 11:28:22.000000000","tz":0},"subject":"Set octavia_amp_image_owner_id by default","message":"Set octavia_amp_image_owner_id by default\n\nThis patch sets the octavia_amp_image_owner_id default so that the\nOctavia configuration file has amp_image_owner_id populated.\nThat is important security setting for production deployments that helps\nto avoid faked images being used\n\nChange-Id: I22c56f32d7308803e9363f9375d7f6206ccecd41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/788f51cac6f26227fdcdc262b963d96a16286e5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/788f51cac6f26227fdcdc262b963d96a16286e5e"}]},"branch":"refs/heads/master"},"9ef492e3b2663d035061745b23b471f3c8d5da27":{"kind":"REWORK","_number":3,"created":"2021-04-23 13:45:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/787670/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_octavia","ref":"refs/changes/70/787670/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/70/787670/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/70/787670/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/70/787670/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_octavia refs/changes/70/787670/3"}}},"commit":{"parents":[{"commit":"370ea30b99e57c711ede72d90fd2323385d4a2b1","subject":"Merge \"Remove unused handlers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/370ea30b99e57c711ede72d90fd2323385d4a2b1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-23 07:13:41.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-23 13:45:09.000000000","tz":0},"subject":"Set octavia_amp_image_owner_id by default","message":"Set octavia_amp_image_owner_id by default\n\nThis patch sets the octavia_amp_image_owner_id default so that the\nOctavia configuration file has amp_image_owner_id populated.\nThat is important security setting for production deployments that helps\nto avoid faked images being used\n\nChange-Id: I22c56f32d7308803e9363f9375d7f6206ccecd41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9ef492e3b2663d035061745b23b471f3c8d5da27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_octavia/commit/9ef492e3b2663d035061745b23b471f3c8d5da27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
